)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~727066","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~stable%2Ftrain~Ieed29584c3b8cbb7f827a1f482f2860d5015a93b","project":"openstack/openstack-ansible-os_tempest","branch":"stable/train","hashtags":[],"change_id":"Ieed29584c3b8cbb7f827a1f482f2860d5015a93b","subject":"Pin telemetry-plugin-tempest to 0.4.0","status":"MERGED","created":"2020-05-12 06:20:02.000000000","updated":"2020-05-12 18:56:35.000000000","submitted":"2020-05-12 18:55:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727066-1589309703809-df443003","meta_rev_id":"c28f595fc235c8b44f1128498622ebee67f53554","_number":727066,"virtual_id_number":727066,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 18:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-05-12 07:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2020-05-12 17:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2020-05-12 17:04:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 07:30:25.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-12 17:04:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-05-12 18:55:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2209481d9e11fbb54fb63a598c04b55128ddfe48","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 06:20:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70e17688d105074cee6aedb80fbd16f7e8d93908","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-12 07:30:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f3f378613cefb266339402ee04b4ac0f97b70d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:13:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9d7385568b034a5e8c148d91aefc0f09 : SUCCESS in 8m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07d7222b468c4f46b3c3e404d9c33ef3 : SUCCESS in 1h 13m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6781338c542d478ea77b31aa42fad82a : SUCCESS in 1h 22m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/55c5c1acc69f4ce6bf83dd641d33e263 : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9f0d27f2f9c94217b7f951d1ef02bd80 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/744d8afa23734fccb866214e1cafe3d1 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ed7aa8eeaf774e4b8b0092dd40b52702 : SUCCESS in 2h 49m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/945a7c8ece2940228b14bc530879a539 : SUCCESS in 1h 52m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d44219e0c7744c2a89ac9055c1f83ee7 : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4cabffa58180448f90c9de8141a4b170 : FAILURE in 34m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0be37403b2e44538f1d593eef1d6927 : SUCCESS in 1h 05m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deae2e4a7614440e8311ce53fb7e8c69 : SUCCESS in 7m 43s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/e7dfc70ebd2f4d05b6caff824dab2450 : SUCCESS in 1h 16m 32s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/591d8736187842c6970a797a71e64ae6 : SUCCESS in 1h 34m 33s","accounts_in_message":[],"_revision_number":1},{"id":"71fe6bcf5ae8dc58efffc53ac8ad19c366f1a8cb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-12 17:04:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"beaa4cedf7a8c9f880c53017a4fada616389a786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:04:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8cbb3f6f5266b8cdcb58bfac5619ee19bc63d3a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:55:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2700c41ceb2d43f78ec69e0da2d67eba : SUCCESS in 13m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3f2c7b54a26f4d388f9f489b7282a8da : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/bb6da17c90db4136954a41fc544a5a07 : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/905a0b0f05164d6b9fa7f5bdb5f6ba7f : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bca39787436b44efb8f768cde9ba8b71 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef3565be9afe4477a3ec933325c2a286 : SUCCESS in 1h 12m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/893fb74c5036445a8f14509a47162508 : SUCCESS in 1h 46m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5029b4c5d574e708f9a406e0dc1b8fe : SUCCESS in 1h 08m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ee220a422e249c280b6da7f805a43c3 : SUCCESS in 1h 00m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8311efc4ce7543f8baccdb2c9e4f56e0 : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/ad61fc9eb8fd41d1916113df25b85843 : SUCCESS in 1h 15m 21s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/7dadd9b0ee024d3583a13898ff97bbf7 : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e9c253c7f4396b8f4c82eeb193065186be08ef89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:55:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c28f595fc235c8b44f1128498622ebee67f53554","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:56:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab5ba2c23cb49bab722478f89670c56 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8490ada0044abf0c5103dfeb80bf1f069941ee1e","revisions":{"8490ada0044abf0c5103dfeb80bf1f069941ee1e":{"kind":"REWORK","_number":1,"created":"2020-05-12 06:20:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/66/727066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/66/727066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/66/727066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/66/727066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/66/727066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/66/727066/1"}}},"commit":{"parents":[{"commit":"2c1f6fe262c462c3235e5ba84862eec7214e72e0","subject":"Make upgrade jobs voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2c1f6fe262c462c3235e5ba84862eec7214e72e0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-12 06:17:09.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-12 06:19:40.000000000","tz":60},"subject":"Pin telemetry-plugin-tempest to 0.4.0","message":"Pin telemetry-plugin-tempest to 0.4.0\n\nThis is the latest version that supports python \u003c 3.6 so this pin\nis required for debian stable and centos7 jobs\n\nChange-Id: Ieed29584c3b8cbb7f827a1f482f2860d5015a93b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8490ada0044abf0c5103dfeb80bf1f069941ee1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8490ada0044abf0c5103dfeb80bf1f069941ee1e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
