)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~746848","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~Ieae36a1d85a7da84ee2cd982a20dcabe9e65511f","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"fileperms","hashtags":[],"change_id":"Ieae36a1d85a7da84ee2cd982a20dcabe9e65511f","subject":"Set mode for copy operation","status":"MERGED","created":"2020-08-19 07:49:45.000000000","updated":"2020-08-22 18:14:13.000000000","submitted":"2020-08-22 18:13:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746848-1598119981195-1fceb40a","meta_rev_id":"bda0ba587365af67badfa2450c83b836950991ea","_number":746848,"virtual_id_number":746848,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"date":"2020-08-22 13:19:26.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-22 18:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-08-20 07:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-21 07:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 07:42:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":27379,"name":"jrosser","inactive":true},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 07:50:06.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-08-19 07:50:13.000000000","updated_by":{"_account_id":27379,"name":"jrosser","inactive":true},"reviewer":{"_account_id":27379,"name":"jrosser","inactive":true},"state":"REVIEWER"},{"updated":"2020-08-19 07:50:33.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2020-08-20 07:06:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-21 07:42:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-22 18:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"395429b3d79675928c01bd0dc6cd0b99e5d21650","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-19 07:49:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455321f0e5d5e42eb3da06f951f5e6a74e27ceb9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-19 10:09:16.000000000","message":"Patch Set 1:\n\nbut the default _is_ 0600, and you set it the same. Are you sure you didn\u0027t mean to set it to 0644 or smth?","accounts_in_message":[],"_revision_number":1},{"id":"77600c0abdfb3038a58b3a2f54f6b008c99970ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 10:12:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2ef217f277a841439b0f0bd098c93fba : SUCCESS in 7m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bb70f3e7d8d04fcda9dbad0b6cb2a0a5 : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d0e938f798f4dacb29b22aa5c7a6311 : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7b34ca16b9f34724a491610e4094130d : SUCCESS in 1h 17m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cf037a4fcfd4044bf47d3f1b1ab9724 : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1577152bd854423a1ea036d8ef33221 : SUCCESS in 1h 08m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/afa08f5daae443c294c46e87f69b98f3 : SUCCESS in 2h 14m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c981ea8916df402d8f533988ff374c17 : SUCCESS in 1h 52m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/412b3d5cc73d4b1181e7a184e353ac96 : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e9b5a2b36d5547a6a7045b8fe24e51d6 : SUCCESS in 1h 02m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a7ba803204248fea7d37dd0eded4ac7 : SUCCESS in 55m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3206271ae2b54415a1a35bf72a5160e0 : SUCCESS in 51m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afabe3b4f2f4a56bde82f9d21140f25 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/e70eae60f9a94977907a33bd0c3f8a3d : SUCCESS in 1h 04m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/723c12cedb174f45835e0f9b88ad7aab : SUCCESS in 6m 14s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9f417b296624443b85ae3a9b937665d8 : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"18dea4b028a8731d278bff9029e727e215317deb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-19 10:19:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"703d3bf7a4d752b4ad52a3ca55392d7112f8061e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-19 10:19:42.000000000","message":"Patch Set 1:\n\n\u003e but the default _is_ 0600, and you set it the same. Are you sure\n \u003e you didn\u0027t mean to set it to 0644 or smth?\n\nYes, correct I need to set it to 0644, fixed in new patch set","accounts_in_message":[],"_revision_number":1},{"id":"80986015b148430c2a5aa235ea35bbdab77f185f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-19 10:35:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bfac46833c14386311e14e6ac9aed46dbb9b91d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 12:12:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ab882953c38240e18d9cdc39911d0696 : SUCCESS in 10m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/62ceeed746e247ed9314071ca01db136 : FAILURE in 58m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b83998000eb2416e895446c616ad753a : FAILURE in 1h 12m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/580a9730a024408bb7763924956f41c3 : FAILURE in 1h 09m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54eccaefbf5842459e039144edabfc87 : FAILURE in 56m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cc0c65253344dd98c9e2ef373c4c88e : FAILURE in 53m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4ce658ad777543c5ac93197f33e28745 : FAILURE in 1h 45m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7aee586e63fd4b909024fe0eaf847ee6 : FAILURE in 1h 29m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4a9dc874f1334784b6f0ec670c52a376 : SUCCESS in 55m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eb9cb94502f24efd85286be143d0874d : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62a5e60b1103430fa71e9fc913ebdd60 : SUCCESS in 58m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dbf6cb5410a41d8b3925b6600b359d5 : SUCCESS in 39m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409556990db84697add5f0a20c87a27a : SUCCESS in 6m 33s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/831484a317624d30be111634f0251997 : SUCCESS in 1h 02m 17s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fa390a503b10411ca47dfb253343cc54 : SUCCESS in 5m 16s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/36aa66813c134eada0b4f928853d7059 : FAILURE in 1h 06m 25s","accounts_in_message":[],"_revision_number":2},{"id":"046fbbc36ab9095138d1f4a4ef88fe819562d36e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-19 17:02:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"36b1799abcff870a30e8cddbd0989eb308309f90","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-20 05:51:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d9f660c82b8fd738ea90bea7b76feefa67e5dce","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-20 06:50:35.000000000","message":"Patch Set 3:\n\n\u003e (3 comments)\n\nSorry missed the quotes","accounts_in_message":[],"_revision_number":3},{"id":"0a998277180896a6967ea5e07791aaecdf1f1153","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-20 07:06:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d1633e2e44add7de188a77c0fac61517bf457ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 08:28:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6cb21ac5da544bdb804331bf7c05bf18 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d7a560b8390f48328d5bc23e83dfca97 : SUCCESS in 59m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6c343dfd0698408bb4baa761082f8b55 : SUCCESS in 1h 33m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1f941788a2804e96960d58a51d6d8d89 : SUCCESS in 55m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98086da6424e4fa381137714ec1162fb : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdceea42633c4a699d27f6cc3625fedd : SUCCESS in 1h 02m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c270630fe8dc4b67aa5ea545b411e822 : SUCCESS in 2h 27m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85f3b483ce7f465997268ac065a8a931 : SUCCESS in 1h 40m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/739cd866f6f940d3b20ed138eb0408dd : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/af53dd8e1f634c12bd5b406c3f06bbb0 : RETRY_LIMIT in 3m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/798fa9f7ec254157a1cd0a2dcc5a3706 : SUCCESS in 1h 03m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c37b05a568244651ac0ba2b7a2ee9f49 : SUCCESS in 49m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb6418a903db4de79beaf14281c1227a : SUCCESS in 4m 56s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/198d9019947040449b7a9c7eff303d1f : SUCCESS in 1h 01m 24s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0c0be60ff0054be6956ba640f88fc8d3 : SUCCESS in 5m 03s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/50e0ed43ded64f228a004c04e1e0a717 : RETRY_LIMIT in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"86c04ce694fb1c0575b93ab0574585dc4c71f01c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-20 10:00:02.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"9b095682dbbd8a639d0a2beabf428bd3d58aa4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 12:19:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/19d86802b8d248e2b4256490538cf1d3 : SUCCESS in 7m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/67bcd79651b54fda89f6a2e04215cd03 : SUCCESS in 1h 03m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/71a2cdce693e48afa554b3f43c7a9eec : SUCCESS in 54m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a09d6715a26540a0811e440ae3509cba : SUCCESS in 53m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71116cdbb9fa47f5b0c6353c0daf8967 : SUCCESS in 52m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3ec1d72f6647b0a45bb8e9616393de : SUCCESS in 49m 27s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f92a03ec4e864a3bb6daf3f367875e1f : SUCCESS in 2h 13m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15c624c774dc4ac594fd0a6f6cc233d9 : SUCCESS in 1h 50m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7b890c9dc1374cbaa46d4f817f4b51da : SUCCESS in 1h 01m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6d9fb60aea724511aa0fd768ec7f99a7 : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7deaa1b77e75477788a03a3de240b3e5 : SUCCESS in 1h 21m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d01d2daebd340d6bf10ac4cb83ca690 : SUCCESS in 59m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd87b84d87ee425fb92de89041e4c6a8 : SUCCESS in 6m 42s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/190fc58a019d43a5a6b016c8dca2d35e : SUCCESS in 1h 02m 13s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/05666d0f97fb42b5b4f891dcad64169b : SUCCESS in 5m 27s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/89220f5c1add45d7b1ac5102b1b38ead : SUCCESS in 1h 42m 02s","accounts_in_message":[],"_revision_number":3},{"id":"586a4c6109f987f2aaa3195ed7a42f22c66b84df","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-21 07:42:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"81d462414b0fbf4c09d2496295b71e248c77f73c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 07:43:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1247c100a4c00c9153417a8d632088a0acee89e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:22:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/151fbc1718374d6eb5715aef8849d634 : SUCCESS in 7m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9ff5c77e367946f6a123e01395576a11 : SUCCESS in 1h 05m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c6c93dee0374272b8b1fdc0d1b87df3 : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6f058878199d4a58a93f101171d01fe1 : SUCCESS in 1h 16m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f15084c788964882a005f7e603abc6b2 : SUCCESS in 1h 08m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cb43b03fd6f41869e8ccc597b81d7fb : SUCCESS in 1h 05m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/79ae0a0b0d924c08a8b60c008c6c5594 : SUCCESS in 1h 04m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bb2fb5f8d4454d70ab070e217b48aa55 : SUCCESS in 1h 07m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5618db0772ca4e0c98a909fe71740849 : SUCCESS in 54m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d380b65afae6402bb8d27d314840f9d7 : SUCCESS in 46m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53dbb9f16a674e6baf68914186098984 : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/cf081b8ca03141c9aa7e00b47f1cd5d6 : SUCCESS in 1h 00m 11s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4f7d7ae2fe1f479d85f916e75b88e77c : FAILURE in 1h 09m 38s","accounts_in_message":[],"_revision_number":3},{"id":"06517457a29e87e83defea7213c91b27d2322ef5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-21 09:27:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"53a8bf4cdaf5d35ac6458d7ee76956b474091a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 11:48:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1b7b842b709447c2aebb8e5f199dbf12 : SUCCESS in 8m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/abd87ecb1bac47a39e2680f5bbaa5144 : FAILURE in 1h 16m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b22b88cac6284060a6f7bd0db74d8ab5 : SUCCESS in 1h 27m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/28a94c181bbd4a5bacbe5d80390f77c3 : FAILURE in 23m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d679fb8a0224fa789414750c1025ef9 : SUCCESS in 1h 40m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c74431a717554d5bb88fdf37ce00cad6 : SUCCESS in 1h 06m 54s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/67bbe475b8ca4319aca957f81b8253a8 : FAILURE in 1h 06m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b7d9c13054447e59e0cd3dab759ae5c : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f62d254b2c94aee80de1d033e7b6b54 : FAILURE in 1h 14m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/49342c7ef1574f57a57baabc1cc6aad2 : SUCCESS in 57m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff5bd8db151f422e80771cdd290e23f4 : SUCCESS in 58m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e52f68f959d74b57bce8f1e3dee5653b : SUCCESS in 51m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702d2f2d85304e1b8a9cd4ef438e897c : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/f321356f6ddb40eea5ab50bfdf3e18a9 : SUCCESS in 1h 20m 12s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2a03ea4325c641278397cd8c8a0b3f49 : SUCCESS in 5m 48s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0e65990884b7477a8ff4eb8928d6cfd2 : SUCCESS in 2h 13m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2e58c533f175adb33d4b240b7816448d7077f018","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-21 15:41:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ffd30045ed537520368d3613ce002d2f4fbf57d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 18:23:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b7d8bee346db4a18b2fae06f90b1d84b : SUCCESS in 9m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a11ecdec177b479099e8daf6680788b3 : SUCCESS in 1h 12m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/575bfb0280634f5889a8ada4439dbc66 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/610cf182abd048999a179981eec432a3 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bd0683686a0456697e778a7cc6c607d : SUCCESS in 1h 11m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0df0889ff5e49ffba943e5da1558f09 : SUCCESS in 58m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/acf21458502c484c8d1fe9ecacaf6446 : SUCCESS in 2h 26m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f411d9532424786b368d36d8358968c : SUCCESS in 2h 09m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a3ddf12075924f3ba9b45c1937fea0c6 : SUCCESS in 1h 02m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/629c133f22d24574a574335716fb0447 : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24d96f870f164723b22b9161124fb23b : SUCCESS in 59m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ab5cde4c70a430293057c4a5449e1f0 : SUCCESS in 53m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f529de6ad7bb4b8f90fe0afec383d6cf : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/efd1ddd447564728a815d0839e2a23e8 : SUCCESS in 1h 02m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f38abd8a37c64f9cad1a9c5782fc915e : SUCCESS in 5m 59s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/08fc7a908a164488b6a6b55c01bcaf1d : SUCCESS in 1h 44m 03s","accounts_in_message":[],"_revision_number":3},{"id":"deb55da793e6bedb0f2e8e44ace955800254f0e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 18:23:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"11a57e85efb9de8963b93b87f64d5e88bf2c75f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 20:06:58.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/77b7448f1db14899bc09f56abccd9ec6 : SUCCESS in 8m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/df45c5a3715341d4b762e292f061eee9 : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a05e81e80b0448c6a7aa96a49199f1a4 : SUCCESS in 1h 37m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d44d98b1c4d4a618cce10761e480ea7 : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8b5159065784d4f88f5c9dd12516e0f : FAILURE in 22m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79efe45179904c4c8ce29bd23052ec07 : SUCCESS in 1h 10m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7d2687118481440bb586c4eaa32f403b : SUCCESS in 45m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7c06684995e1460fbf351456632d436e : SUCCESS in 1h 07m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1405d080f524066823b33deb0090ed4 : FAILURE in 21m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e97c18056d34e99b960143a978ea7b5 : SUCCESS in 54m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffefeb8e37d14b18949f2782a147fa9c : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/397b80fab7a842a681fac9a806a4cee4 : SUCCESS in 1h 07m 04s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4310ac576621432db40d8d07b6b8c9d2 : FAILURE in 1h 12m 54s","accounts_in_message":[],"_revision_number":3},{"id":"880e6fea02ae9a66960928669ca88661b6263498","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-22 13:19:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e7da2445d41cc70d489aefcaadee5bc56cf2b74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 16:32:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/487d3bb1f3d643a494395286519116c7 : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1fd66a4f6ced47678bfb1a6cdae7edf4 : SUCCESS in 59m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a76fb8f6d834e53b4e968b534151c94 : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/04db0e70578c4f629b1f2ff311ae788c : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba8a335665e846c1bbe4c02b60ae6938 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/814af36f93f34243adcbdeaf5e9e8d88 : SUCCESS in 1h 13m 07s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/604c7a40581c42fa868c0e4cc5f00331 : TIMED_OUT in 3h 09m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d97692e2971349efa0329fdfc6fdd771 : SUCCESS in 1h 53m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2a32e5736b3a4e6a8b4cb5f0d2edc647 : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f6f3c7fdac3640c0b650bc2eb9b2ae83 : SUCCESS in 56m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7f11818744c420d8a9ec7e08999cac0 : SUCCESS in 1h 00m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9dcbb37e6de425cbe8cf00e53dbeeda : SUCCESS in 1h 01m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b572e0842e445099debe024b56af248 : SUCCESS in 8m 03s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/f9652dc034be4913ac95d5a11d01bc3f : SUCCESS in 1h 11m 57s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8590a2e3a5434e84ae9005212c5e876c : SUCCESS in 5m 00s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4e649711fd4b41cf813e08ab4824d43d : SUCCESS in 1h 41m 38s","accounts_in_message":[],"_revision_number":3},{"id":"09bd06c5fdef7c51e0379338713ec99f52fd9de1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 16:32:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9a7ad9fa5436b604ed61c43fa0c54e05b56c69fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 18:13:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b94518f9d48c4931a8fc472462eb8663 : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f319e9703ee24395a763f294dcc088a5 : SUCCESS in 1h 03m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0dbe557d2dd245e6984edf61024a0e96 : SUCCESS in 1h 28m 39s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b248b41188a442c883bd36ed479b9477 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a87d67fafe6b4b2db07898c28a768cf3 : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0f78cedea6441e78495f43776323d9e : SUCCESS in 1h 04m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/622afd00876d443ab76e15212b5038cd : SUCCESS in 59m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0d664f4560ae46f4937b534af519d5ef : SUCCESS in 56m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec0e8fec78de4f80b11dab4461dac475 : SUCCESS in 53m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4c362cb9b4c403685a167fd65b0b17d : SUCCESS in 46m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69256c7bfea646aa89cc9dd51655ae2d : SUCCESS in 4m 17s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/a67b3dc6e07d43a7b6d78b9d6165aae5 : SUCCESS in 1h 05m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a5b01dd578754dd399698a076a024375 : SUCCESS in 1h 35m 13s","accounts_in_message":[],"_revision_number":3},{"id":"20b265a141c3bc288055139a994117c392c9faf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 18:13:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f365d3aeffc3e74796dc61d335703425c24eea7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-22 18:14:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ea951065554db685037ee4fe7ded8c : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3e590ab572417b8f15c680ce483d5b5ed9f12c1a","revisions":{"62b22d55c0cbcaa4b8ff14d657cf170234645d02":{"kind":"REWORK","_number":1,"created":"2020-08-19 07:49:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/48/746848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/48/746848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/1"}}},"commit":{"parents":[{"commit":"0546a91d6a2578a47869f9c875e0450bd593682e","subject":"Merge \"Remove deprecated scenario image option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0546a91d6a2578a47869f9c875e0450bd593682e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-08-19 07:44:23.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-08-19 07:49:30.000000000","tz":330},"subject":"Set mode for copy operation","message":"Set mode for copy operation\n\nWe need to set mode for copy operation otherwise ansible-2.9.12\ncan lead to too restrictive default permissions[1].\n\nSetting mode for copy operations handles the warning.\n\n[1.] https://github.com/ansible/ansible/issues/67794\n\nChange-Id: Ieae36a1d85a7da84ee2cd982a20dcabe9e65511f\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/62b22d55c0cbcaa4b8ff14d657cf170234645d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/62b22d55c0cbcaa4b8ff14d657cf170234645d02"}]},"branch":"refs/heads/master"},"ed8b4d80c9eeaa7d987dcffb7f653d90cb8a3083":{"kind":"REWORK","_number":2,"created":"2020-08-19 10:19:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/48/746848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/48/746848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/2"}}},"commit":{"parents":[{"commit":"0546a91d6a2578a47869f9c875e0450bd593682e","subject":"Merge \"Remove deprecated scenario image option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0546a91d6a2578a47869f9c875e0450bd593682e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-08-19 07:44:23.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-08-19 10:18:44.000000000","tz":330},"subject":"Set mode for copy operation","message":"Set mode for copy operation\n\nWe need to set mode for copy operation otherwise ansible-2.9.12\ncan lead to too restrictive default permissions[1].\n\nSetting mode for copy operations handles the warning.\n\n[1.] https://github.com/ansible/ansible/issues/67794\n\nChange-Id: Ieae36a1d85a7da84ee2cd982a20dcabe9e65511f\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ed8b4d80c9eeaa7d987dcffb7f653d90cb8a3083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ed8b4d80c9eeaa7d987dcffb7f653d90cb8a3083"}]},"branch":"refs/heads/master"},"3e590ab572417b8f15c680ce483d5b5ed9f12c1a":{"kind":"REWORK","_number":3,"created":"2020-08-20 05:51:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/48/746848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/48/746848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/48/746848/3"}}},"commit":{"parents":[{"commit":"0546a91d6a2578a47869f9c875e0450bd593682e","subject":"Merge \"Remove deprecated scenario image option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0546a91d6a2578a47869f9c875e0450bd593682e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-08-19 07:44:23.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-08-20 05:51:07.000000000","tz":330},"subject":"Set mode for copy operation","message":"Set mode for copy operation\n\nWe need to set mode for copy operation otherwise ansible-2.9.12\ncan lead to too restrictive default permissions[1].\n\nSetting mode for copy operations handles the warning.\n\n[1.] https://github.com/ansible/ansible/issues/67794\n\nChange-Id: Ieae36a1d85a7da84ee2cd982a20dcabe9e65511f\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3e590ab572417b8f15c680ce483d5b5ed9f12c1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3e590ab572417b8f15c680ce483d5b5ed9f12c1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
