)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~777309","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~stable%2Fstein~I021015709c8254ff5936ad08d61e1e02ab58c44f","project":"openstack/openstack-ansible-os_tempest","branch":"stable/stein","topic":"add_provider_network_option","hashtags":[],"change_id":"I021015709c8254ff5936ad08d61e1e02ab58c44f","subject":"Add provider_physical_network option to private network","status":"MERGED","created":"2021-02-24 08:14:50.000000000","updated":"2021-04-22 11:24:11.000000000","submitted":"2021-04-22 11:22:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777309-add_provider_network_option","meta_rev_id":"84e9ff37c79b936bb1f32b051b86fa6555a4c050","_number":777309,"virtual_id_number":777309,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 11:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-10 15:55:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-04-19 18:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-04-22 11:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 08:14:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-02-24 09:36:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 08:22:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-03-02 15:56:30.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-03-10 15:55:35.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-04-19 18:12:25.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"f42f1af60b427574e71e250a4be7674f4082bc5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-02-24 08:14:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72a1e004172ee151b6d7de8c273b3f2166e91049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:36:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/db649408c06f4f2c89068f6a991e01f4 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/424c440f4d45490785b994a261e4abc0 : SUCCESS in 1h 17m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7ea8507ec5ca40148ce562655301538b : SUCCESS in 1h 03m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db33ab7525c048f8bdf2fac1b45e987c : SUCCESS in 55m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4379e9547ce747f1a15ab6903ac026ec : SUCCESS in 53m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/933392e64f3046ddb52c78f0b135a832 : SUCCESS in 44m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ca03d9706945e9ab4052e42e9a2edd : SUCCESS in 4m 44s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/cb77d0d8499f46eaba4aa6dfb23516a0 : SUCCESS in 53m 49s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/534004fbedc743e5bd5774000d86d353 : FAILURE in 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"903938d7a7b9516fd21ef9718c16578e5faa7d24","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 08:22:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"364bca32f79d64451c39ca232bc7812a6917f149","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-03-02 08:45:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20c262fc6d720619317ae3d835cac042854761fc","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-03-02 08:46:24.000000000","message":"Patch Set 1:\n\nThe failure on centos 7 is a known issue, and it\u0027s not related to this patch","accounts_in_message":[],"_revision_number":1},{"id":"7214b1499ecbe4c8b81f54eefc01e471452a83b4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-02 15:56:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41bebfa024b5d0892dfad20282101370cda39e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:56:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c32664323a6b89a6c4fbae2b6f26f2a8576907f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 17:08:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4ac038a900e84d20ba2bf04bd012b7da : SUCCESS in 8m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/aabd024857264881a61b1dfb223d29f5 : FAILURE in 1h 04m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ebcecc77701747d080c55d3ef7b00e0b : FAILURE in 1h 00m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a17a81136e1b444b973fb21aec8b0ef3 : FAILURE in 49m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b604c97a634f4d10acee0e2a1a7acb97 : SUCCESS in 53m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b87f5ef069147c7b931daf7279ddb04 : SUCCESS in 43m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6eab1e564204de9a837c1b6515fb62d : SUCCESS in 7m 26s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/72c4e7cb179f45399f8b21e07dfaf75b : SUCCESS in 59m 02s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/f9120439576b44eeb95c2c759d607a42 : FAILURE in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82655ef3bc8b56222a85f6055f230c2b6f134750","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 17:22:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36b052a4bc4ed5d5de953d663c87c0c9121dc393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:47:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/57c96b4f3dd4441bb7368c4b6fef9652 : SUCCESS in 8m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c1d1b443f1e4420e8d496bd22c6e474c : FAILURE in 1h 21m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/910b53f6eeb442cfb065bf9aa3d105ac : FAILURE in 1h 01m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1ce60ec5ac746aba0e41e6052197353 : FAILURE in 46m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f30f7f7b6c246fc93d65f8694aff47c : SUCCESS in 52m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23cff615172f4a548d06e8ecf704c154 : SUCCESS in 48m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86564bd9c95c42c4a6fd88c8c5f51753 : SUCCESS in 5m 09s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/c376f43a6ad14f5591683a836eac3243 : SUCCESS in 1h 03m 21s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/9d18cf5132a44bda82f4476c8f3ebbaf : FAILURE in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e23fb44273207e2be327991fbca2d8c807d8babe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-02 21:13:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1feceebc55c927b4ae3f9b415482a788ee08c415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:22:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dabad205cc0f4facab5eba16f54d6162 : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/90693b93f7f7425a96241a69342812f7 : FAILURE in 57m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28622ee85698444c98f3becd599905a3 : FAILURE in 1h 07m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11ddf63a1caa486e86d53124181a4e72 : FAILURE in 54m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/25e2a6d11a0641e0b486f38009a83868 : SUCCESS in 52m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac3c564df010464e9792d3c7d2372c26 : SUCCESS in 55m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed1440f93f1407a9ee1672fcb9de4e0 : SUCCESS in 4m 49s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/be182f18e4e8429eac6b7dcbdb452f66 : SUCCESS in 54m 03s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/4a972f0295db40efa28bb3471080f570 : FAILURE in 22m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d147dbf285f3e5dd0894c4055048c2a86fc3fa9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-05 13:19:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6c5fef6c9e213062f253a7cd4a183c15cd2695e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:45:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ab500e1a8cab4bfcb0a130f180252db0 : SUCCESS in 6m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e8d2adfd1c334501ae760eca3ff78ece : FAILURE in 59m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/77b81424eeb44ba18829a5147b260167 : FAILURE in 58m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/881d1101375f4a7f8ec911fc833d92d8 : FAILURE in 45m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e88f61ff4664f0985a6327609672de6 : SUCCESS in 52m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92737025a1a748ac8778f2012fabe631 : SUCCESS in 42m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3c23d820d0f469ebec57a9ab3d0717b : SUCCESS in 4m 34s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/c44ffcc7e33548ff891f32e9df3a716e : SUCCESS in 54m 49s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/825bec7529f049d0a6fe0598e6ee52b8 : SUCCESS in 1h 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e76bfb8188c62e3ba233f76b368223c5c66237a0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-08 08:30:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6a8b6e7ffe8374e320e2166aacc35dcf01cf9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:46:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6d6d63436d0d4936bc3e0ee5e7a9a50a : FAILURE in 4m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f80b990a7f5d49d5b5a4855448533b63 : FAILURE in 59m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f02390d6b9c343b583467db7e26a504d : FAILURE in 1h 10m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fcddcc563614cc5a6bbcf9b54daa2e8 : FAILURE in 53m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d64663b272549b187f05d26c64d6d76 : SUCCESS in 59m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ace23846eaf14560abb3a0d7ee1532d7 : SUCCESS in 48m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4602d8071c04e7f9755db81af025647 : SUCCESS in 5m 34s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/d73af173c410431ea09dcaeb1bf08ce9 : SUCCESS in 59m 42s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/2e776c63b7c24d0e8e984859264e08a9 : SUCCESS in 1h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbe9733fb1fa567f75702927c932b8d59d8d1ba6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-09 08:21:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"07dafaa96d3d45c3aa6eec3d26b5eb01dde541f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 09:39:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6ed5bd24eca04de7af7205fd16339cb8 : FAILURE in 5m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a2dc0e02f9d14ba9aa5cf3bd098d1722 : FAILURE in 57m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8727399f30b941d3b7ef9171318bb6fd : FAILURE in 1h 12m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7283618e9210477c8e875ae0614e7201 : FAILURE in 50m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/477b37ec5c6e438682510cf831746c24 : SUCCESS in 54m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9804176e848c4739ac9c5a59f75c15b9 : SUCCESS in 46m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153b3fc057ae4c11835fb6f1504151ba : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/a501db49de6949a4b33a2d4fc2831357 : SUCCESS in 56m 50s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/bbf753069d1a48b984b39412c80c4537 : SUCCESS in 1h 09m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d23b41b2e0b4f690990b62b763d61329053ee21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 08:00:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"74cbb74c25130936c6509f4c291a85c119d5bcfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 08:10:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"64e8ac44bb5f1fb7ed80e301315f00c8865d2c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 08:19:36.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"dafe174c83127bb04ef66f3812a09bc8d6ae6526","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-10 11:40:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2369f6cc0a67954f423993089e3686f2904f74f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:14:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/68479605019f4d7cb0ab507bf6282414 : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5fb8858f49a649db9be7b59e326161a8 : SUCCESS in 1h 06m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/37aa594c636f4de59e0c99605b4eba3c : SUCCESS in 1h 11m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae37d41ec36d4c04a883482e15680c2e : SUCCESS in 52m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/76e0d8bd532d4ca5ac7688edf88d629f : SUCCESS in 47m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f657f235f7e40c9b78a81dd0e5f934b : SUCCESS in 48m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758e1f0a0b244400affb0a0de3ad7815 : SUCCESS in 4m 32s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/cdb4432c0ac44f0ab0966f904025c33c : SUCCESS in 58m 38s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/1f45c1b144b6426e9763e582d7e4ed6a : SUCCESS in 1h 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"765b099f3d61cbeac3abce5dc42da5e4a4b08be4","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-10 15:55:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"329eb8844819b32136defc28b2f164c1405babfc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 18:12:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4bda54c547b5dbea6343128c74faed1308eb651d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:49:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cf4999510aab3f6a258b26ddd9826b9a5d34ea70","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-22 11:09:51.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"7d24b2f817fffbd7790f3fb2c9e5ff6c42f428a7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-22 11:09:55.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8fb86993bafa4c6890698207d52261721181e5b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:22:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ec7d45ed6dd74c22b7fb83af55175ec3 : SUCCESS in 8m 40s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0fd12d37bdc444f0b176e64c4f91480b : SUCCESS in 55m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b1c44dcfcd8248a382c56ea13c5c71eb : SUCCESS in 1h 14m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/111e0f90769746a385c1ee25d1a4b3f1 : SUCCESS in 54m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c5aa9e0890fe4f3bb5279ebbf6dad7ab : SUCCESS in 48m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ba0dd455dba4eeaa5e500f9310669a1 : SUCCESS in 39m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d4d319bf9b4f6f968ec4c95e6111f4 : SUCCESS in 4m 51s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-7 https://zuul.opendev.org/t/openstack/build/65e42b7608f64a2e809e8cf4d83ff3d5 : SUCCESS in 59m 08s\n- tripleo-ci-centos-7-standalone-os-tempest https://zuul.opendev.org/t/openstack/build/2fa4ba26f9454888af6704e48f1d8702 : SUCCESS in 1h 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"809d5f44b3afa4133b937c2983175840df0d9e88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:22:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"84e9ff37c79b936bb1f32b051b86fa6555a4c050","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:24:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd222f1cafac4cfd885dcfbcc108f47f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"baac187a60a3fa19150bb1d1757857b6e2fd1be1","revisions":{"9bcd02da8f7e0985951711e33b76b86ef0d5e8b8":{"kind":"REWORK","_number":1,"created":"2021-02-24 08:14:50.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/09/777309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/09/777309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/1"}}},"commit":{"parents":[{"commit":"cfbfd94b3df42470f5a8aa85f9644a261f2b3bed","subject":"Pin telemetry-tempest-plugin to 0.4.0 for python2 compatibility.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/cfbfd94b3df42470f5a8aa85f9644a261f2b3bed"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-08-22 07:50:00.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-02-24 08:14:18.000000000","tz":60},"subject":"Add provider_physical_network option to private network","message":"Add provider_physical_network option to private network\n\nThe vlan network provider, requires a physical network type, otherwise,\nit will fail\n\nConflicts:\n    tasks/tempest_resources.yml\n\nChange-Id: I021015709c8254ff5936ad08d61e1e02ab58c44f\n(cherry picked from commit 24969898114dffc792bfc23a15f60875f4a647cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/9bcd02da8f7e0985951711e33b76b86ef0d5e8b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/9bcd02da8f7e0985951711e33b76b86ef0d5e8b8"}]},"branch":"refs/heads/stable/stein"},"4f335209e69e27d84f5a3548ca6d661b3eadfc30":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-10 08:00:22.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/09/777309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/09/777309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/2"}}},"commit":{"parents":[{"commit":"5bd1a8f9b95a574818ff5deb5e8602f3fb02c2a1","subject":"Pin cinder and keystone tempest plugins to versions compatible with python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5bd1a8f9b95a574818ff5deb5e8602f3fb02c2a1"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-08-22 07:50:00.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 08:00:22.000000000","tz":0},"subject":"Add provider_physical_network option to private network","message":"Add provider_physical_network option to private network\n\nThe vlan network provider, requires a physical network type, otherwise,\nit will fail\n\nConflicts:\n    tasks/tempest_resources.yml\n\nChange-Id: I021015709c8254ff5936ad08d61e1e02ab58c44f\n(cherry picked from commit 24969898114dffc792bfc23a15f60875f4a647cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/4f335209e69e27d84f5a3548ca6d661b3eadfc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/4f335209e69e27d84f5a3548ca6d661b3eadfc30"}]},"branch":"refs/heads/stable/stein","description":"Rebase"},"baac187a60a3fa19150bb1d1757857b6e2fd1be1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-10 08:10:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/09/777309/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/09/777309/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/09/777309/3"}}},"commit":{"parents":[{"commit":"5bd1a8f9b95a574818ff5deb5e8602f3fb02c2a1","subject":"Pin cinder and keystone tempest plugins to versions compatible with python2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5bd1a8f9b95a574818ff5deb5e8602f3fb02c2a1"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-08-22 07:50:00.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-10 08:10:35.000000000","tz":0},"subject":"Add provider_physical_network option to private network","message":"Add provider_physical_network option to private network\n\nThe vlan network provider, requires a physical network type, otherwise,\nit will fail\n\nConflicts:\n    tasks/tempest_resources.yml\n\nDepends-On: https://review.opendev.org/779595\nChange-Id: I021015709c8254ff5936ad08d61e1e02ab58c44f\n(cherry picked from commit 24969898114dffc792bfc23a15f60875f4a647cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/baac187a60a3fa19150bb1d1757857b6e2fd1be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/baac187a60a3fa19150bb1d1757857b6e2fd1be1"}]},"branch":"refs/heads/stable/stein","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
