)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~783637","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~Ieb050fe65947e3d2da1beb5062f509036e32a7da","project":"openstack/openstack-ansible-os_tempest","branch":"master","hashtags":[],"change_id":"Ieb050fe65947e3d2da1beb5062f509036e32a7da","subject":"Log tempestconf inside tempest_log_dir","status":"MERGED","created":"2021-03-29 11:12:17.000000000","updated":"2021-04-20 14:26:47.000000000","submitted":"2021-04-20 14:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783637","meta_rev_id":"424cd9faeff125e6045f40731724babad979bb62","_number":783637,"virtual_id_number":783637,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 14:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-04-20 14:26:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-20 14:26:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":1,"date":"2021-04-20 14:26:47.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-04-19 17:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 13:52:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 07:39:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-04-13 18:01:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-19 09:08:40.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"e2c880a771491080e2f06eea3edda70d47b6569d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-29 11:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7561e41cb748ff68cd2a5000ae344bfc2320f23a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 13:52:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/78b4a6e2369d442ab8ee42f8aeccec9a : SUCCESS in 11m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/00580fa86cab4d379bf8cf9d1ff32850 : FAILURE in 34m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/75e408fd7a7a42f3ab6935f7845ea540 : FAILURE in 1h 05m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a1b0d1a76fe4c5098ba2ebbbf2dbcb6 : FAILURE in 1h 00m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39e3c90126604626b324215fb922efbc : SUCCESS in 1h 23m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0213218c5bfa4bc693b0a06be9e4affa : FAILURE in 1h 34m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46fb544f8f424f5badc7bfed12ff541d : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/230acf212f024b5790a9edef6ee5a52d : SUCCESS in 47m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/813e61e4be534e2d8a8a8ffa9f826c30 : SUCCESS in 47m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b01968a04ec4cb6aae4d527a4183954 : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8 https://zuul.opendev.org/t/openstack/build/a404af3a06664e16a440a079732f9133 : FAILURE in 51m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5af7b4612a3943fd8b34abc61e3229d8 : SUCCESS in 2h 20m 31s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/1bc2599f2cac479f83b2fe09848d047b : SUCCESS in 51m 36s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/ba90cc3bbab04ab58bc45cf1e042ce0e : SUCCESS in 2h 23m 15s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b36e08207bca45afbc1ddeb77193bd61 : FAILURE in 5m 22s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f38296633b0047f5ac4ba0a545a031b0 : SUCCESS in 1h 22m 12s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/4be12f9aa81747f9821c44f305de3614 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":1},{"id":"33f8a1ffb339ff8bf5d523add0c5c6a98cbe3891","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 07:39:28.000000000","message":"Patch Set 1:\n\nSo, tempestconf fails with CERTIFICATE_VERIFY_FAILED while connecting to the keystone. So we need to make it use insecure or reach internal endpoint instead of the public one","accounts_in_message":[],"_revision_number":1},{"id":"35126a121d47087b953b2465562e497caebedacf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 07:44:26.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/openstack-ansible-os_tempest/+/769966","accounts_in_message":[],"_revision_number":1},{"id":"b640f56d88340a2d5c8993b72db16b3627d75647","tag":"autogenerated:gerrit:restore","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-01 15:17:39.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"19a6a998c7b76ee91f12750e34ceacb4a36fe1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-01 15:18:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"04a342a10aa1a750a86effb3c7caf79f77aca83f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-08 09:19:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8e61b70027c2c2d8e8e204ab12178ab5d39237e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:05:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d50064fc8d594b7185243863e2cce520 : SUCCESS in 9m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7aea9b0b2d394315a72a1e4ec6f563ed : SUCCESS in 1h 08m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b4b4b00f7a24470ab5d3ec33eb7da239 : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f7fbf643cb248ed912b3462436fb9ab : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0607fca920b4df5a01743101cd3ad5e : SUCCESS in 1h 06m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ff0f807db9a4456e83fef63618fa1cb7 : SUCCESS in 2h 01m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d430085f35f4ec9a8cc4cebb5355468 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/154eb0b18e714a22818595ce89978a4a : SUCCESS in 1h 15m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/799a365db29d49908184bf7da068e2af : SUCCESS in 57m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b278d9932a543d9979e7abc3c740c7e : SUCCESS in 6m 43s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/2655903c71124f34b826626a42d04af7 : FAILURE in 45m 56s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9333f90374334652977d27396ca420c1 : SUCCESS in 2h 40m 57s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/37668d33c1bf4fcea3ce48aea4070ceb : SUCCESS in 1h 05m 32s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/2ff14a86d6074aa19ef6811d9402e995 : SUCCESS in 2h 36m 06s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/629a71b057f74e19a5d4fdec07e93a68 : FAILURE in 6m 03s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/27f87e3948824a99b9f9418e35ac77dc : SUCCESS in 1h 31m 58s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/dc6ca2693ede4b9e97d391e7097775e6 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d63894a8a7bc7b617729dee2a89db520f327843a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-13 18:01:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25adc0e70e46c2220721e4e8549c72eb9815ee4b","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-04-14 09:29:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"34d4b378ac7ad503f81a62b138fd1a37bf5ee574","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-04-19 09:08:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f838bbeaa54e98e2454c955c1a3f5ac996f588b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:08:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f0d104c8e2c0c8091435fb97633d45eb4729052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:29:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8a2dba9bad0d423ab2b50be674ab046d : SUCCESS in 12m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/239ff48943564443ab3e14428f4c729f : SUCCESS in 1h 17m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ec65e3dad634e14b158b7db81645c1a : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f069a6a27b3e48cbb52d76035e04dc8f : SUCCESS in 1h 16m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d525576e73714ab598f93dcd48b2ad50 : SUCCESS in 1h 03m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e578a9fca9429bb6372e08e898f1a0 : SUCCESS in 57m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8dc8027a5ac4c5b9dc09e2578151982 : POST_FAILURE in 8m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/60b253f5091943d18ce6a364f4aaddc1 : SUCCESS in 2h 19m 07s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2fd6c24dd26b4e57b8c024e338e84ba8 : SUCCESS in 1h 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b85cfda4593b6ce0cfcd4c158e253c62df9f2b70","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-19 11:51:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a311f420f572a93a380e20a3289178047a225f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 14:57:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1c031c6072864ff69f93bdf8335f837b : SUCCESS in 12m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/96d2bc3ed91a4562adad51ad35a7e123 : SUCCESS in 1h 01m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4c6f3218dfd84b8b838dcc9439d5c4a3 : SUCCESS in 59m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41ee48aac7bc49779c4831a0074b6da4 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34999ed8cb6b4225b0af16a067abbfc8 : SUCCESS in 1h 16m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/33559e33d3f44e9c815ac56adccc120b : SUCCESS in 1h 49m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6bf135f7904400ea6edc92cb5ecaa13 : TIMED_OUT in 3h 01m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a993222bfd14efb9cc1ea9cba6e1b46 : SUCCESS in 1h 00m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62bef23f8546493ca81b1062d87187d1 : SUCCESS in 59m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab6eaeb1e3c7414aa23febbd58b3f0fb : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec56821eb2ec4967973b91d0690b8991 : FAILURE in 42m 34s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4f67d2877914d1a8e2de3027c1d0fc3 : SUCCESS in 2h 23m 43s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a547d3ebbecd4d4d8544abf8c546d1c4 : SUCCESS in 55m 00s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/af61100d1cb543739e1b4437c370d20c : SUCCESS in 1h 40m 02s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0a547d525f274bf8a134fcd0fe9dc4e3 : FAILURE in 5m 00s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1771d0173ba8485c8030023832c692fc : SUCCESS in 1h 19m 32s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/cdd4b9fc65aa48b0ac114d479678656f : SUCCESS in 48m 05s","accounts_in_message":[],"_revision_number":2},{"id":"97a789924ece74f359e1e1a79f144f8e6271e7a2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 16:03:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6c77696ca36beb4739c0c44581b18f6b9c6e28cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 17:56:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"70918f009f4fbe0037a3b8fda6892ff9155c12ef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-19 17:57:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db51d5f95099d899ac503fc80e9f918e44d7309c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 21:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3a06c751cea4d29a7600f0f82e1f880 : SUCCESS in 9m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/77d7f10bc36644ecb2767a33c252f01b : SUCCESS in 1h 09m 34s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/461a813bd4594ed4bbb975ee4f373246 : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/426033e76f99421a9cae30dcf0ab68ae : SUCCESS in 59m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/768776a16fe3438bb0b698b08c8e67e4 : SUCCESS in 53m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/015ea8c7c58c4a9d86517b1c3a3c04e7 : SUCCESS in 2h 10m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5555c94934714596b2a9e13460fbe48f : SUCCESS in 1h 47m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/20bea0f8a3e5404aabca1af06022e6ac : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f683cb803ae4d57a226f169a02ccc83 : SUCCESS in 53m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06925ac0d3774b00b702f2424cb188d8 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/505ec9163d8249e1b35313cafb1cfa31 : FAILURE in 54m 10s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e64740afe7c84a8c9ec3dd1c2e6ab850 : SUCCESS in 2h 58m 33s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3d1ecf7ffab9482b8870b0e08346421a : SUCCESS in 1h 10m 37s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/45f0c76d88f146edbb923c56a98de481 : SUCCESS in 2h 40m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4d354228e8224be4a6aed37b13268038 : FAILURE in 6m 09s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ac95a20930c54e52a1da2e4e04a65502 : SUCCESS in 1h 32m 55s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/b79529875cd34ddfb8e7378fbca4be55 : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":3},{"id":"1ec2f935829664406f4b6c094aae018d3b8c3ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 21:03:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"137fc752ff816bbed3525e67bdeeee6c1b39959f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 23:51:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/94842e30d95e48209ac231e2c33798e9 : SUCCESS in 9m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/20afa7efe264436094be90358c56064f : SUCCESS in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94ee64d251ce4dae8bad0d557e4c9c67 : SUCCESS in 1h 20m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3dcc28cb3d14f21807a4f1c64abb27d : POST_FAILURE in 1h 04m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fad4bbc40dde4aa2bde9f15aee879490 : SUCCESS in 59m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1200f4b834074146b7c0c0d6ecbb8ab0 : SUCCESS in 55m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60026eb8517344889594aed2c7350014 : SUCCESS in 5m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6dbd55dd96364dcf893a219319210596 : SUCCESS in 2h 43m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0328e3cd3bd940528075a8cd705caec3 : SUCCESS in 1h 33m 07s","accounts_in_message":[],"_revision_number":3},{"id":"233488e1874be4bd0f950f877c9e47b963a915ae","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-20 07:44:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fe6b9c5bcbd503924672331f10a96ca51f3257c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:09:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6ecdead81c8c43e08604d342e6fc2eac : SUCCESS in 12m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f7e6200fadc94a8eb0184edb6ce0f0ba : SUCCESS in 1h 11m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7cb36be97f864b77a67d1e5d24a76e09 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21f0fa05e91d448dbfc75b57ee9da679 : SUCCESS in 1h 14m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18627f2efcc248fd986a324d298d3bac : SUCCESS in 1h 04m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a2d0bf0119f4cf2a233b51696d78a32 : SUCCESS in 1h 45m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/375cf704958148809353d47f05d37852 : SUCCESS in 1h 40m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/061e9333ecd346dc981c53dc6251435e : SUCCESS in 57m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6f93977a75e446db2217c63b84812a8 : SUCCESS in 59m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a10a58f43142deb81bd16477726d1a : SUCCESS in 5m 53s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/eed1190978ff4104a804452ac429cdd9 : FAILURE in 54m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/78171f3ef419427ab0707c2790fbfc1d : SUCCESS in 3h 17m 17s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ac68aeab9cb649dbb709f2349971b8a6 : SUCCESS in 1h 25m 21s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/639aa254f86d4450b970d29ebbbd5496 : SUCCESS in 2h 57m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5333925a171a42f39f89cec7f31143f6 : FAILURE in 5m 13s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8b5d4d3ce2b14154a2ec7e4087828133 : SUCCESS in 1h 46m 50s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/46926c80796e4a5cb60ef287c1087317 : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f2da2b99d4d6b90557d5c09d199ce0d18a78a5a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:09:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ead0aaf1a7cfba585aa5b1b44dd268e1b622a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:25:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d6650ba906b34adab4abe8b963701b18 : SUCCESS in 10m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/349e0d5ba5f94e4ea42fe4ea64136544 : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37571056730e447ea2bbe91a5af8d538 : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11f8188666c6456e9c1b52fc4c9b49ec : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6979256415f14df8b8a5088a2cfac373 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a9a649ce1fc4c9e930f73866e47ce2e : SUCCESS in 1h 00m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23029f6f1244c4880229b71a3a902f5 : SUCCESS in 6m 48s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/24335e9159d44db99faba7e63e02c794 : SUCCESS in 3h 09m 24s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/199be28f31994a06a1c00cd4d8e23ad7 : SUCCESS in 1h 39m 48s","accounts_in_message":[],"_revision_number":3},{"id":"9d75125a989d6ea2ea0a5992e58e0e8900b05699","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:25:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2eb5b56643cd28c9db3d836e4d88e9e74761820d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:26:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b1f27306c343fabc74463b2bf672ba : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"989557bff3e09f1675f737171be43b8a819539b8","revisions":{"7048614e660bd655dd11a365ac136fe5191a9548":{"kind":"REWORK","_number":1,"created":"2021-03-29 11:12:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/783637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/37/783637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/1"}}},"commit":{"parents":[{"commit":"f08b2b1d4eaa00464c5795318d175c867f8880e6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f08b2b1d4eaa00464c5795318d175c867f8880e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-29 11:12:06.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-29 11:12:06.000000000","tz":180},"subject":"Log tempestconf inside tempest_log_dir","message":"Log tempestconf inside tempest_log_dir\n\nThis will allow us to collect tempestconf.log by CI\n\nChange-Id: Ieb050fe65947e3d2da1beb5062f509036e32a7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/7048614e660bd655dd11a365ac136fe5191a9548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/7048614e660bd655dd11a365ac136fe5191a9548"}]},"branch":"refs/heads/master"},"54deb930449fc204c57c259a5eafb3fa083fcb0a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-01 15:18:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/783637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/37/783637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/2"}}},"commit":{"parents":[{"commit":"53892664aba6b64be106e18d44e24755e49784e9","subject":"Switch tempestconf centos-8-stream nodes and set as NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/53892664aba6b64be106e18d44e24755e49784e9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-29 11:12:06.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-01 15:18:00.000000000","tz":0},"subject":"Log tempestconf inside tempest_log_dir","message":"Log tempestconf inside tempest_log_dir\n\nThis will allow us to collect tempestconf.log by CI\n\nChange-Id: Ieb050fe65947e3d2da1beb5062f509036e32a7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/54deb930449fc204c57c259a5eafb3fa083fcb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/54deb930449fc204c57c259a5eafb3fa083fcb0a"}]},"branch":"refs/heads/master","description":"Rebase"},"989557bff3e09f1675f737171be43b8a819539b8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-19 17:56:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/783637/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/37/783637/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/37/783637/3"}}},"commit":{"parents":[{"commit":"e6df466ebc90183932f650c62800c3d716518496","subject":"Merge \"Add trove tempest support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e6df466ebc90183932f650c62800c3d716518496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-29 11:12:06.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-19 17:56:57.000000000","tz":0},"subject":"Log tempestconf inside tempest_log_dir","message":"Log tempestconf inside tempest_log_dir\n\nThis will allow us to collect tempestconf.log by CI\n\nChange-Id: Ieb050fe65947e3d2da1beb5062f509036e32a7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/989557bff3e09f1675f737171be43b8a819539b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/989557bff3e09f1675f737171be43b8a819539b8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
