)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~783814","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~Ide5368b9c26b81a2824fef350ee9d9dd692cdf18","project":"openstack/openstack-ansible-os_tempest","branch":"master","hashtags":[],"change_id":"Ide5368b9c26b81a2824fef350ee9d9dd692cdf18","subject":"Fix u-c filter regex","status":"MERGED","created":"2021-03-30 09:16:07.000000000","updated":"2021-04-19 12:48:05.000000000","submitted":"2021-04-19 12:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783814","meta_rev_id":"967159791961ef1dd3b3bb3107f36a553b24ec0e","_number":783814,"virtual_id_number":783814,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-19 12:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-04-19 09:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-04-19 12:48:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-04-19 09:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 09:38:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-03-30 11:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 09:09:50.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b6d2708dc5b56e9cd7392526a5f5938a0a6a1c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-30 09:16:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9afc5c41df753a84a0d781e31a47d3835f309033","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-30 09:38:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b19cd20bbc91dadfc0115d9cc0d4e351d758d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 11:52:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b4fa23a0e1fb4d688ddf562394c1c2c7 : SUCCESS in 9m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c007c7ff760f4777bcf39b91687bf760 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/89450c4bf2e743218e094b9e724de338 : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00124efb124642d9a040e81e63780af7 : SUCCESS in 1h 00m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/251c1ed8243d4bacb4ecbddae5bc204c : SUCCESS in 1h 21m 09s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c7e8c2a9bdf241b78896cb32c429bdb3 : SUCCESS in 1h 25m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83923dad6b404de5b02898ed6ba72cfc : SUCCESS in 1h 08m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1733bb8af52542eea2b0afb767e0579b : SUCCESS in 53m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c49e59bd5851483c9a21d2a8c730cfa8 : SUCCESS in 59m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/641d1cb78c0c46369e6521886c6c55d0 : SUCCESS in 7m 15s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8 https://zuul.opendev.org/t/openstack/build/8345a773430c49fdb39fd937c807c980 : FAILURE in 53m 41s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4dc0ad58bc204daa9096cbef70352ad6 : SUCCESS in 2h 18m 13s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b49ae1a109dc42ef87dd57bde03ad55a : SUCCESS in 57m 18s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/453789dca4a94b3b8a4222274d5f0a15 : SUCCESS in 2h 20m 14s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/de9e3dbeeefb457a91d2b87a49134f1d : FAILURE in 5m 25s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fd69f9b4e0a442299a2f3c27bc85449d : SUCCESS in 1h 18m 41s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/ccc285e8288c48b9bcf61753e23725ec : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"42789797f005063826973181e5585d240f6976fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-02 09:35:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8a09f1743f60ec4269e2d1978e83926e5bdcedd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 11:56:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bb7d4a0786da4bfaa7a0615e4436b5f5 : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07ebf25499b84a308367915a3f2f497b : SUCCESS in 1h 13m 54s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f230d349bcd34266b216b4b99bc18eb8 : SUCCESS in 1h 14m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8229c710a6f54f64aa44fff367b28604 : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/524af3aa30054fd1973fce061d617dd9 : SUCCESS in 1h 11m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0701e482e39e4860ae0706f3daa13b10 : SUCCESS in 1h 46m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b67e9829339a43248ab17bed8d35641c : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6634edbe32cb490488428fa11ae1fd8d : SUCCESS in 57m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df590ecca7f341f699a1a8cbae35914e : SUCCESS in 55m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0263cbc88cb44b289df189c56a0b8993 : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3821a24a4ad438fa2304babeb370b52 : FAILURE in 48m 39s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a3357fdef175441a8fa0c9d7256ee938 : FAILURE in 42m 15s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/27a9da4468e34bf5a3c2fbfe78864131 : FAILURE in 42m 36s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/033a6bb8833147c2b6fa35bf5e7fae75 : SUCCESS in 2h 17m 59s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c172b8ca97a746a693980f5492f1fe96 : FAILURE in 4m 50s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/545464466bf54f5689d2cf08122b1c63 : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":2},{"id":"614ffd26bc38b34dbdf25002e9802ad02495bc39","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-08 09:19:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2f3232483153c9ca8875e254582496d041195ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 12:19:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/56d30d8da6e049d8a90aeb2cdadd2937 : SUCCESS in 10m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4078328bae8944289762124a7d842a88 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dba14339e4624f64aa238d284d51faf7 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e86194481fe94f079c57fb182276ebba : SUCCESS in 1h 22m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f17539359764caea01feb27d50dad92 : SUCCESS in 1h 16m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d8d64881d46d476c8085c473b8238f4d : SUCCESS in 1h 51m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9927ffc4cb14986a44f95f4801e6dbd : SUCCESS in 1h 46m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e625118d67d4dc6ae3c2fbfaf1b0ec7 : SUCCESS in 59m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50a7586dcf5546e4af07a04e0c279d5b : FAILURE in 37m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eee4e7b4e7e427bb9bd682680b04100 : SUCCESS in 6m 08s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2f4a43b08f844a08985e0d6a4e9273b : FAILURE in 52m 29s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d8fb15c579654909a13fc14be0c48781 : SUCCESS in 2h 33m 09s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/098a323804ee44d084f5860112fb8e44 : SUCCESS in 46m 36s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/c243d733672345778b2799c27c780098 : SUCCESS in 2h 53m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/785156d55f2e49218735d6c2ac2fc1cc : FAILURE in 5m 56s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/847e083d06f948ff85d7a5156a6bce27 : SUCCESS in 1h 23m 51s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/7f5c7e9e446146649120ba7c5fce479d : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5d30cf2b4439fe95651ac95c20567bd23b97c760","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-14 21:32:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ec09320e5d5406a7256ee71f50bbc57688fc5523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 00:07:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7cd5d94da7ed4901a4dd522ba3d2015c : SUCCESS in 9m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9c0d95ab65e4476cbbd03328e3d9b57c : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b61b58a8c91d4c2bb81c558cef1d5bfa : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b54e9a7e74d34926833e02661a8a6641 : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fc3359ebf5a408bb712e779f77c80b5 : SUCCESS in 1h 41m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/52e47f7166c040aea76a898ad1f705c3 : SUCCESS in 1h 44m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7176deba9b74393a9b52c5e34eee3e4 : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4ed60308f906411bb820631bf426e778 : SUCCESS in 55m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef24c3d87d6748ac9dc57183bd19aa06 : SUCCESS in 1h 10m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e6ef2a9b804aacb9f130ddf12d5238 : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/4945a6aa72e34445a86debeb1330edb6 : FAILURE in 50m 45s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6166517bd3a34fc184d27e8be66abc01 : SUCCESS in 2h 24m 20s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f3114f2e373f453a8cb2fd4596d3f16b : SUCCESS in 1h 10m 56s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/e49fb5e2adfb43e0b47db952999f19a1 : SUCCESS in 2h 31m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/96832c9586e34435be90328404526041 : FAILURE in 5m 00s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0ad3f1bc169942b0a0594b6380ed72d7 : SUCCESS in 1h 16m 54s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/9f0dd7f5642844da9e1e20d223094a4f : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a65e419ae4ae67d22706b64c215bcbd5b88f51fe","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-04-19 09:09:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d85b2f86c9357aa585354b36fd107bb5605237e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:10:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45941bc099fd9bedbcdd6b3112cd07b5675179a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:46:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6a984265efa94524aee1aa95dd35a5e5 : SUCCESS in 12m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7b59015ea7f04228a45854d6782ddb79 : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90b56f41472943bab8ea799f07592592 : SUCCESS in 1h 18m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/686c50a89c47440c8a2b1b9fc63fdb12 : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0023ab82605344a085d7556f14147d2c : SUCCESS in 1h 02m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71f051aafa5a4a7dad6b397120c530b3 : SUCCESS in 1h 01m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7ca52ef0db41e1bc6950ed2d403431 : SUCCESS in 5m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bde8e36ef42a43cbaf3c0adfaa7c7787 : SUCCESS in 2h 59m 24s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ca98afb2d2f24a2b8729b605fc12547f : SUCCESS in 1h 39m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b83f2b7bb9d07631a882b45e03b124847cb5f078","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:46:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b4ecf76b95528158aafdbc063f5b71cdd8f240be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 12:48:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59256530147743ed8f5bdaa6b8763a6f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"435fc4115c3d7ce2161c573b6af9063b75c31707","revisions":{"f34c9bacd8cb0f64293219f151434b4858ef17a6":{"kind":"REWORK","_number":1,"created":"2021-03-30 09:16:07.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/783814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/14/783814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/1"}}},"commit":{"parents":[{"commit":"f08b2b1d4eaa00464c5795318d175c867f8880e6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f08b2b1d4eaa00464c5795318d175c867f8880e6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-30 09:15:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-30 09:15:50.000000000","tz":60},"subject":"Fix u-c filter regex","message":"Fix u-c filter regex\n\nChange-Id: Ide5368b9c26b81a2824fef350ee9d9dd692cdf18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f34c9bacd8cb0f64293219f151434b4858ef17a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f34c9bacd8cb0f64293219f151434b4858ef17a6"}]},"branch":"refs/heads/master"},"435fc4115c3d7ce2161c573b6af9063b75c31707":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-02 09:35:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/783814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/14/783814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/14/783814/2"}}},"commit":{"parents":[{"commit":"53892664aba6b64be106e18d44e24755e49784e9","subject":"Switch tempestconf centos-8-stream nodes and set as NV","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/53892664aba6b64be106e18d44e24755e49784e9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-30 09:15:50.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-02 09:35:49.000000000","tz":0},"subject":"Fix u-c filter regex","message":"Fix u-c filter regex\n\nChange-Id: Ide5368b9c26b81a2824fef350ee9d9dd692cdf18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/435fc4115c3d7ce2161c573b6af9063b75c31707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/435fc4115c3d7ce2161c573b6af9063b75c31707"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
