)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~784019","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I29ecf06fc625ef0d93424ce8e04bb592cb2ce07b","project":"openstack/openstack-ansible-os_tempest","branch":"master","hashtags":[],"change_id":"I29ecf06fc625ef0d93424ce8e04bb592cb2ce07b","subject":"Switch tempestconf centos-8-stream nodes and set as NV","status":"MERGED","created":"2021-03-31 07:29:58.000000000","updated":"2021-04-16 22:31:17.000000000","submitted":"2021-04-16 22:27:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784019","meta_rev_id":"0fa8a2d545312185b20bfcf088184a3611954d2d","_number":784019,"virtual_id_number":784019,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 22:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2021-04-15 16:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2021-04-05 12:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2021-04-15 16:27:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 10:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-02 09:35:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-04-05 12:19:41.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-04-15 16:27:11.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"}],"messages":[{"id":"f852e5059883ec6fecfeae8f43a51915cb0da47b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-31 07:29:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0abe81f929838c51a1d52b7d35fef2034f22a041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 10:21:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/044d90b1513142a6a274c6cab7deb35d : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cea3e56a826c4872840d77f63b1df500 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d300e5f6620544f6b65b8cb0c805b67b : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8d09b70834042f98541838bafbc8487 : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fab63f4d2e44fb99342f8e16678df32 : SUCCESS in 1h 11m 05s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/116f0e41f679426789f6fc419318cf23 : SUCCESS in 1h 51m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4064741849294538be17c4e5ee025d67 : SUCCESS in 1h 35m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a656aff4bb0f4bf8be467e65c1022f01 : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90eea0d1f1e84d5fa6a742ced91a1f3e : SUCCESS in 57m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e12b5ce4d91b4e6f83b4ded3c8a6318e : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9628de5ba7d4ad284c3f63fb22add21 : FAILURE in 48m 51s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e3e2a2b30d934d859df0a98052921331 : SUCCESS in 2h 37m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d9ec7a710dd746fdad79d298047404b2 : SUCCESS in 1h 06m 21s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/a876220de996423483828ae8883bf387 : SUCCESS in 2h 19m 00s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c606675fc3a04a99b60dc1499899801c : FAILURE in 8m 48s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c6a5a5dff2ca44ee99f48bec25e19abb : SUCCESS in 1h 28m 27s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/a7f05d4a29f049ac89dd3c724fe3eb6a : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"829fc03ab61d50b33caf8db4346f3d9d1e552d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-01 15:17:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eb61649b2bd1e3bfd744c0ecaafec6e93e8519e9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-02 09:35:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"64bf0ccd51c6469ca962392851b2ff6cbf940eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 12:01:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/222fae48db10430a88f00fdf2945efae : SUCCESS in 10m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e53688dd2bd149258e049a3402b90e7f : FAILURE in 20m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d1aef963d8d34058bdc2880819092cbe : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f41d5469d43406a986ad814c0f0807a : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e4f3093615c459e92e137e9f6ceca3f : SUCCESS in 1h 14m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e534f2fa8aa94df6a01c8cd7f0bc21fb : SUCCESS in 1h 40m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22eb10e9ae374bd39b7eae8eaf1487af : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e917f5e20dc24fc1a215d175986bf3bf : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11070b803617484fa4c76f46ba689fa2 : SUCCESS in 40m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40b1fc22baf4817a1223d940548b818 : SUCCESS in 6m 25s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/59d0515048bd4c7b95df6e08498a87bc : FAILURE in 56m 49s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a544d8b690a84d87b06a18af3493908a : FAILURE in 49m 31s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4ba85a017dad4a86828947c403cf691f : FAILURE in 54m 14s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/bfe69581cc1341a992b9d72c78f55f71 : SUCCESS in 2h 20m 57s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c22eb778d4374672ba437d1aa41ea8b6 : FAILURE in 4m 41s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/2ef5a3813aba45f4b8a348c32ed58939 : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":2},{"id":"00485306d72ad55ff9ac0819de781ae97437e7b8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-05 07:45:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"93812e974695d9e348c6210c284ab32960cc9d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 10:11:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/160c2f00495441eeb13f20a5c50cd744 : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8e417eb691be4ae8a53d48c4a694be7f : SUCCESS in 1h 12m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f51627daaf024d7cb13713c9f8024a1b : SUCCESS in 1h 16m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8d0489e0c0f45ab8eaa76acb3f846be : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87d7b580d6924089b9c514cb7707bdd7 : SUCCESS in 52m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8ec7b310ba5a44f8817bf4b093a48842 : SUCCESS in 1h 44m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeae7be0235c4f2bb512d2ff6385e672 : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa6a3009ce0a4a1e9031f498aa77c58f : SUCCESS in 54m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3b85a4827244ee683b342b81d207d3a : SUCCESS in 44m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0df4e26ec034a6aa97993a57eaa2576 : SUCCESS in 6m 15s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/b22cff5fbb0448debe71abfdbe6928d1 : FAILURE in 37m 59s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e57eef6350184b0a8e2f48b7919bd479 : SUCCESS in 2h 19m 32s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3cddef3755c847a18fae5e077c9b8116 : SUCCESS in 44m 41s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/d4c0b2b2c65d4104876ae37f9d90e125 : SUCCESS in 2h 22m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/9f49db0fdf4c425a8774d80463808e01 : FAILURE in 5m 23s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/04d279edac8e476ea7401b2aa1c80dc4 : SUCCESS in 1h 19m 48s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/0c9ff330b0bd42e7bd1a2877235e63a4 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":2},{"id":"0f45f69453177d2997a5f8fbcea6ad65c0360276","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-05 12:19:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"568de80184519656698173808acda9a35ad4789a","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-04-15 16:27:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c40870ed09215aa286ebe40ac9b6f41ee61ba8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:27:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a2d3b0027f9b31d9ca8cb89a63ad589eb5cd640a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 19:08:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4dd89933c584c1595a89b301f7c6a21 : FAILURE in 10m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ff4b29cc36e6430892774cfb52452fa8 : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a009a97367646faaed06abd1e390820 : SUCCESS in 1h 30m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e16a6fc35694146abba88933b4b92ec : SUCCESS in 1h 22m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a18157e2f60472dac9f68a255eae22e : SUCCESS in 1h 07m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca573788be7a4775a95e00682164e847 : SUCCESS in 43m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b4f2e16c91e4ccab7f78035f4b35a5d : SUCCESS in 1h 01m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25295b862e84dc5801a0e6c2f1ed1c7 : SUCCESS in 6m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/32d743fa04b740dbbc559385efbf7f21 : SUCCESS in 2h 34m 11s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a16710a6286f44e1b056c6ea4c7a8d9d : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f8d880d46713dcef05f6e65f73b6c11e4cd842bf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-16 16:36:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"088b766395b66ff7d0b457b65358bff8acb237b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 19:53:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f4d03b09aaf3423f84a0811e51890876 : SUCCESS in 10m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ec86a364702a43018e36e4312a0d7d96 : SUCCESS in 1h 06m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/05393f42a9a14df08ffe5104e4351bca : SUCCESS in 1h 16m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87238b5e33e149ff824b3d63b363c733 : SUCCESS in 1h 05m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b1452f9f671436c933b73042678fa72 : SUCCESS in 1h 13m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e1e09696fbe47f1be08b85dd88552aa : SUCCESS in 1h 57m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26a2f20ca0f6463681a62365c3e5f086 : SUCCESS in 1h 45m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/be2eea71d24b4f8c852bd3cc31d5d204 : SUCCESS in 1h 02m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76f5c85e25ea4135941b6e95ab227ce7 : SUCCESS in 59m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b16eec1ce3b4c7dbbb843be4a1a9f67 : SUCCESS in 5m 55s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef538698d0ae410d84eac9113567498b : FAILURE in 1h 00m 48s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/acb7b38b51e84c43b8ed9917d5157ef0 : SUCCESS in 2h 58m 52s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ae01ef9336c546198fef49cd962330a9 : SUCCESS in 1h 06m 34s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/5bffd37e04754564bcf0d45bfd0a7e18 : SUCCESS in 2h 37m 54s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/055547cac3b4463f94ce0c33fcbb921a : FAILURE in 4m 43s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e781a9a670ad42fcaad102f68d748c8f : SUCCESS in 1h 38m 25s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/93b9f8b107254ee5af1cbc71814a7996 : SUCCESS in 1h 02m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e3b31357a26f2e1390069a403952ed2f4e494d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 19:53:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79bbef60ce81a5558080449d8254bd714dac8c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 22:27:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2128e59d72bb43029accc8a0a72e9501 : SUCCESS in 8m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8baf10249fb04961b593afcf740fbaf8 : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/316cd3f26e464315a103e77a3f5a2eec : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6301f1cf612f4895b9f62a7daa871c4d : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/62edd46044404dbea6438e34ecafde52 : SUCCESS in 1h 02m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663cb965ed0b48bbacc161852772c15e : SUCCESS in 1h 07m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36dfad52b9de48d9991902164339eced : SUCCESS in 9m 23s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/69c5cdc1057f41c9a7c4162999fd6e54 : SUCCESS in 2h 32m 30s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e87d8c4b742343b9af4b5df68ff665d0 : SUCCESS in 1h 25m 31s","accounts_in_message":[],"_revision_number":2},{"id":"1e995836ebb80ba51eab99edc8f0dd9da355385b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 22:27:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0fa8a2d545312185b20bfcf088184a3611954d2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 22:31:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fe6006b20c424fa0018d8d46675989 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53892664aba6b64be106e18d44e24755e49784e9","revisions":{"187cd2f88d47d1f0db15710083f57d2f75071d32":{"kind":"REWORK","_number":1,"created":"2021-03-31 07:29:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/19/784019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/19/784019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/1"}}},"commit":{"parents":[{"commit":"f08b2b1d4eaa00464c5795318d175c867f8880e6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f08b2b1d4eaa00464c5795318d175c867f8880e6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-31 07:28:52.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-31 07:29:18.000000000","tz":60},"subject":"Switch tempestconf centos-8 job to centos-8-stream nodess","message":"Switch tempestconf centos-8 job to centos-8-stream nodess\n\nDepends-On: https://review.opendev.org/784009\nChange-Id: I29ecf06fc625ef0d93424ce8e04bb592cb2ce07b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/187cd2f88d47d1f0db15710083f57d2f75071d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/187cd2f88d47d1f0db15710083f57d2f75071d32"}]},"branch":"refs/heads/master"},"53892664aba6b64be106e18d44e24755e49784e9":{"kind":"REWORK","_number":2,"created":"2021-04-01 15:17:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/784019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/19/784019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/19/784019/2"}}},"commit":{"parents":[{"commit":"f08b2b1d4eaa00464c5795318d175c867f8880e6","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f08b2b1d4eaa00464c5795318d175c867f8880e6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-31 07:28:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-01 15:16:04.000000000","tz":0},"subject":"Switch tempestconf centos-8-stream nodes and set as NV","message":"Switch tempestconf centos-8-stream nodes and set as NV\n\nTempestconf has hardcoded [1] usage of publicURL without any option to\nrewrite that. This results in CERTIFICATE_VERIFY_FAILED since we don\u0027t\nhave trusted certificate for public endpoint.\n\nWe mark job as NV till implementation of PKI topic [2]\n\n[1] https://opendev.org/osf/python-tempestconf/src/branch/master/config_tempest/clients.py#L106\n[2] https://review.opendev.org/c/openstack/openstack-ansible-specs/+/758805\n\nDepends-On: https://review.opendev.org/784009\nChange-Id: I29ecf06fc625ef0d93424ce8e04bb592cb2ce07b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/53892664aba6b64be106e18d44e24755e49784e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/53892664aba6b64be106e18d44e24755e49784e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
