)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~813198","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~stable%2Fwallaby~Ifba6836d824e7e6b956a81271d95d6dabe6c4163","project":"openstack/openstack-ansible-os_tempest","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2021-10-12 12:17:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifba6836d824e7e6b956a81271d95d6dabe6c4163","subject":"Pin neutron-tempest-plugin to v1.6.0","status":"MERGED","created":"2021-10-08 14:48:26.000000000","updated":"2021-10-12 12:19:02.000000000","submitted":"2021-10-12 12:17:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813198","meta_rev_id":"ce357421c024010fc3294a2ec1fb3f573d418c38","_number":813198,"virtual_id_number":813198,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 12:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2021-10-11 08:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-12 09:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-12 09:54:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 17:44:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 08:36:44.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-10-12 09:54:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"77d2d73d1f59b5535ca638f93401d5062e2315c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-08 14:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c2afd3dd23b41e5258b78b64cd7948b360a7939","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-08 14:49:21.000000000","message":"Patch Set 1:\n\nAssuming there isn\u0027t an obvious alternative I\u0027ve missed this would need backporting to previous stable releases. I\u0027ve pinned to the last plugin version which worked, but this could equally be changed to use the last release specifically for W.","accounts_in_message":[],"_revision_number":1},{"id":"481be78c19f61f5c9ce5e737fa5ed3d48860d754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 17:44:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1322d91ad12f4faaa92596c5479bdf6d : SUCCESS in 18m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/905f872109264b44b56e514e7e2cdafc : SUCCESS in 1h 33m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/859ecd8268d94801b849e73aebc56c39 : SUCCESS in 56m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/62fb0df2145c409581115b64e96b511f : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/62b6f4df4f0348cb90a1a5fda9039ef3 : SUCCESS in 1h 27m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94c4853faf1f4ab5b0cf5827598203cc : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f91be94ceb5475493f0f1f846c96bc5 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4ae0a00d9fd94d0786ab89d14f7c167f : SUCCESS in 1h 52m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54292d4ef37c4c369c4d6d8eb983b4ac : SUCCESS in 1h 35m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/eff6dc81b1da4bada3410f0b068c1e2c : FAILURE in 43m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c165209f8db4a4cacfdbfdfa09c4d98 : SUCCESS in 58m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ae262d129e432fb1bfcb5688ab7477 : SUCCESS in 7m 28s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/64c55527df684163a4442e6aa50c711c : FAILURE in 46m 01s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/34922333d6b941ed810ad96a58c47d59 : SUCCESS in 2h 41m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7716f04904944b16a8fbafc6e80e31b6 : SUCCESS in 1h 18m 26s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1f5ddb92973249fd8141a57e91c5bd42 : FAILURE in 5m 55s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4e4e3035bfd843859e5c460bafd6d58d : SUCCESS in 1h 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3e57c36a1b447c9b93359ee6b716f2a12fee54e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-10-11 08:23:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"932d2b181ac35c8789b265ecd909a556fd8a5a77","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-11 08:36:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"50d0805064cbc88f072ec64e21487c80dba1be48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:34:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ad22f0b82c94838b328b955aae253fb : SUCCESS in 15m 26s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dbbb99df2dd644459fbb523724005371 : SUCCESS in 1h 06m 29s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8febc04de18b45c7ad84561686a99ef2 : SUCCESS in 1h 14m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/83635c08b999470eba6093ff6dc54df1 : SUCCESS in 1h 16m 39s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b44657dd66b4605b9eb1acbd76dc59f : SUCCESS in 1h 15m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d727092ca1854889a3450dfb26db6540 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfbf38aac8345c59a9adfe7cf9d7bd4 : SUCCESS in 1h 00m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3bafb1cbd5834a93925cea994b8121e4 : SUCCESS in 1h 33m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef264fc817744e2081634c60f15fe1f6 : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b84c6328b3484cdbbeb3dec8cc767470 : FAILURE in 43m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c575f4ace3604c639d3f5567bbc88d95 : SUCCESS in 48m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b42b964ff5b4315a1e5bb7bdfaef210 : SUCCESS in 5m 43s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/9fe5055034a74c2ca56af08ffddb12b3 : FAILURE in 45m 59s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6aeaa0249884411c83491a716bee5bcf : SUCCESS in 2h 07m 09s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c7b8630dc40c4a55bcf74696ed51110e : SUCCESS in 1h 14m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3a3fe13c90ae42b1a210dca245189899 : FAILURE in 4m 50s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6977d7b62b9141bd83c3551a798cbdb9 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3956b2b2589f95e6be848394110b82830b3e9522","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-12 09:54:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e7b04de66302cb97ee7291a238ad10f79d9bd6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:56:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03b6ee40647a27cf2c7b4dd57b9e61e6219445a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:17:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7154c6561070467ebb1ed037306ff5a2 : SUCCESS in 12m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d25dc71c5ef042ff81c0e97765ad2b92 : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a9ad6c69f3ec4f33ba87d5b164e45ccd : SUCCESS in 1h 30m 50s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/24c9d78556ea4217b23169e9993d8405 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/746f10d390254c98aad758d8e17c1e04 : SUCCESS in 1h 25m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce182b0212e74777b3fd74e175bd0949 : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0856e9dd4eb049babf7c4e114db4e8dc : SUCCESS in 55m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791f0a7b9dc140ed983442b9b52ae99b : SUCCESS in 5m 36s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7f87365974a14b74bf1ed3f725c6bbe6 : SUCCESS in 2h 19m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b16cfa4691b341bd83bb52f840273260 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"458cd905f595d83ae6535cc33baee47b0c9533c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:17:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ce357421c024010fc3294a2ec1fb3f573d418c38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 12:19:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694972d0f6374656a7228c7ff5beaaf9 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd399b80e9de12452f1bee620bc60a69c142bf40","revisions":{"a8e01223987d94c7520abcb7cfb8c3395aaae976":{"kind":"REWORK","_number":1,"created":"2021-10-08 14:48:26.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/98/813198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/98/813198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/1"}}},"commit":{"parents":[{"commit":"42f33d8cae55c7e6c361564010e0651703bc3ac5","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/42f33d8cae55c7e6c361564010e0651703bc3ac5"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-10-08 14:45:29.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-10-08 14:45:29.000000000","tz":60},"subject":"Pin neutron-tempest-plugin to v1.6.0","message":"Pin neutron-tempest-plugin to v1.6.0\n\nLater versions depend on a more recent version of os_tempest than\nis used by wallaby or earlier. This requires the\n\u0027qos_limit_bandwidth_rules_client\u0027 component.\n\nChange-Id: Ifba6836d824e7e6b956a81271d95d6dabe6c4163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/a8e01223987d94c7520abcb7cfb8c3395aaae976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/a8e01223987d94c7520abcb7cfb8c3395aaae976"}]},"branch":"refs/heads/stable/wallaby"},"bd399b80e9de12452f1bee620bc60a69c142bf40":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-11 08:23:09.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/98/813198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/98/813198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/98/813198/2"}}},"commit":{"parents":[{"commit":"42f33d8cae55c7e6c361564010e0651703bc3ac5","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/42f33d8cae55c7e6c361564010e0651703bc3ac5"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-10-08 14:45:29.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2021-10-11 08:23:09.000000000","tz":0},"subject":"Pin neutron-tempest-plugin to v1.6.0","message":"Pin neutron-tempest-plugin to v1.6.0\n\nLater versions of the neutron tempest plugin depend on a\nmore recent version of tempest than is used by the OSA\nwallaby release and earlier stable releases. Later versions\nof this plugin require the\n\u0027qos_limit_bandwidth_rules_client\u0027 component.\n\nChange-Id: Ifba6836d824e7e6b956a81271d95d6dabe6c4163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bd399b80e9de12452f1bee620bc60a69c142bf40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bd399b80e9de12452f1bee620bc60a69c142bf40"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
