)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~813990","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I5c05255da78a2513af1a15bea1b0757fc366acd1","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"netstat","attention_set":{},"removed_from_attention_set":{"30742":{"account":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"last_update":"2022-01-07 04:27:49.000000000","reason":"Change was abandoned"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-01-07 04:27:49.000000000","reason":"Change was abandoned"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2021-10-19 12:12:42.000000000","reason":"Chandan Kumar replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-07 04:27:49.000000000","reason":"Change was abandoned"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-01-07 04:27:49.000000000","reason":"Change was abandoned"},"8367":{"account":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"last_update":"2021-10-19 12:13:53.000000000","reason":"Arx Cruz replied on the change"}},"hashtags":[],"change_id":"I5c05255da78a2513af1a15bea1b0757fc366acd1","subject":"[WIP] Need to add netstat logging in the background","status":"ABANDONED","created":"2021-10-14 12:20:51.000000000","updated":"2022-01-07 04:27:49.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3c30b1349d36c88a5d8e1daa4db04e05c8debbd","_number":813990,"virtual_id_number":813990,"owner":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-20 15:55:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":-2,"date":"2021-10-20 13:32:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-14 12:36:48.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-10-14 12:36:48.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-10-14 12:36:48.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-10-14 12:36:48.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-10-14 15:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-18 11:00:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-10-20 13:32:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"dbbf92cff8389d9cd79d9aa5fbd9de811736c721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-14 12:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"880c392ab82e82eeffd5c55126ba70fd70b3680f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 15:36:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98c921acece94f679cb91e190cecf1e2 : FAILURE in 16m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7cbee77bf19a4426bfc94b0fe8299ac6 : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0a56a7866a4044519f724b72fb4168ad : SUCCESS in 1h 22m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a46abacfb32d45449b35e57030dd1223 : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccbfa38cbd244cc78a4b104735368171 : SUCCESS in 1h 40m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2613c313b8c42a2bd24f5f6423d5ae1 : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/249f429f994a491d909f48e16447fcf8 : SUCCESS in 1h 11m 29s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b29a24d780f4b85ba1629db2d4100c0 : SUCCESS in 1h 46m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c0c7825bbb44f24a10fe175b695278b : SUCCESS in 1h 48m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ae0b4b118f64564b8fe37d2b2007432 : FAILURE in 42m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/363622ca9d314f7ba646e9d452d45ccf : SUCCESS in 1h 01m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddf6ab157214d76ba70ae011848f22d : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/139d7f7d8e1247d286c24370ddacf2f3 : FAILURE in 42m 52s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7aacd6f0a18f4f1d9d2e793a6cd48750 : SUCCESS in 2h 48m 00s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7ac5a6b020a14c8da8500ae550560dfe : SUCCESS in 1h 09m 52s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/a04c9363011d491f89ef04d7dbdf1605 : SUCCESS in 2h 56m 11s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/561f4646c9d543aaa6f61ac69f59852b : FAILURE in 5m 16s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/eca88d426aab4c3fa5bb0bbbe3a1ae3c : SUCCESS in 1h 25m 31s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/a80ca475ff8d4217a0d54ffdaba96907 : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"cc1b8aa5f5c48ed9867bd9f807f0ec29469ae4c4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-10-18 11:00:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46c56b25d46031ea7c33fc9d5e9911582ef5cbbc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-10-19 12:12:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"356314c449d22c38f76a60cde73d385a859761cc","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-10-19 12:13:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37f2079d7816f4ee604d594a7007c19b61f87446","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-10-20 13:22:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"340000b73ff8c0aad67f83300c4ff31012402a97","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-10-20 13:32:26.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"468a3be10b0526ae4b7e1fef8087a0c516b9e4b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:55:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0e60b7d317f4945b75f054cce1602c3 : FAILURE in 20m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4ff1e6b2faad420dbed53e916b96030e : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1292cc7593c44c5e9ce7df4489614c47 : SUCCESS in 1h 10m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21e41cc72081408086957a0ec1305299 : SUCCESS in 1h 36m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a857582368646fbbce103ee97c8ede6 : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c341d420f12947cf9d958d50cf50b592 : SUCCESS in 1h 25m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1578b1e2e75841d5a8ef2b71ca5a29bd : SUCCESS in 1h 08m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/abc899c2345646a8bee4d1eb360265cd : SUCCESS in 2h 10m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69590d6c728e48baa326b1ce63216313 : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/fb4cca95f37a4951a2f928c862f7c0cc : FAILURE in 46m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f78152a3a1c54f19bdf6cb6b978829c4 : SUCCESS in 1h 01m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b921acd4d64848a451ff996aafcaea : SUCCESS in 6m 13s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3e58c5c09cc4484959d653909a347da : FAILURE in 43m 30s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6744fc7f25954c8f85c6d0b225e724c7 : SUCCESS in 2h 28m 08s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/90493c27475e4caf933e368248660d8b : SUCCESS in 1h 14m 37s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/e4078151c8964882a9b75f3c593f2442 : SUCCESS in 2h 28m 28s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b5dc4acd54b64016b27a3065da72f344 : FAILURE in 5m 51s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/afdc050926f145c6ad90a13580a60625 : SUCCESS in 1h 15m 51s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/b5b76f26cf9f478689f0c722d443fcc0 : SUCCESS in 1h 10m 01s","accounts_in_message":[],"_revision_number":2},{"id":"d3c30b1349d36c88a5d8e1daa4db04e05c8debbd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-01-07 04:27:49.000000000","message":"Abandoned\n\nabandoning this patch since change is not required","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c36ebeab7c8c1b1eadf7bcf415bc8d7bd22de385","revisions":{"932c94b391ea8f8a4800ae8d75f655c29802b72c":{"kind":"REWORK","_number":1,"created":"2021-10-14 12:20:51.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/90/813990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/90/813990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/1"}}},"commit":{"parents":[{"commit":"f866f7ffeedcd199ea1ed465aa39e11e5ed09c14","subject":"Update cirros version to 0.5.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f866f7ffeedcd199ea1ed465aa39e11e5ed09c14"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-10-14 12:16:13.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-10-14 12:20:47.000000000","tz":330},"subject":"[WIP] Need to add netstat logging in the background","message":"[WIP] Need to add netstat logging in the background\n\nThis patch helps us to log netstat while tempest is running.\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: I5c05255da78a2513af1a15bea1b0757fc366acd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/932c94b391ea8f8a4800ae8d75f655c29802b72c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/932c94b391ea8f8a4800ae8d75f655c29802b72c"}]},"branch":"refs/heads/master"},"c36ebeab7c8c1b1eadf7bcf415bc8d7bd22de385":{"kind":"REWORK","_number":2,"created":"2021-10-20 13:22:02.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/90/813990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/90/813990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/90/813990/2"}}},"commit":{"parents":[{"commit":"f866f7ffeedcd199ea1ed465aa39e11e5ed09c14","subject":"Update cirros version to 0.5.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/f866f7ffeedcd199ea1ed465aa39e11e5ed09c14"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-10-14 12:16:13.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-10-20 12:58:33.000000000","tz":330},"subject":"[WIP] Need to add netstat logging in the background","message":"[WIP] Need to add netstat logging in the background\n\nThis patch helps us to log netstat while tempest is running.\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: I5c05255da78a2513af1a15bea1b0757fc366acd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c36ebeab7c8c1b1eadf7bcf415bc8d7bd22de385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c36ebeab7c8c1b1eadf7bcf415bc8d7bd22de385"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
