)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~825163","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I3f50443b94a5c0e1ebcc566cd268af58561c5e03","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"tempest-damian-2021-12","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-04-14 02:52:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f50443b94a5c0e1ebcc566cd268af58561c5e03","subject":"Rename [orchestration] section to [heat_plugin] in tempest.conf","status":"MERGED","created":"2022-01-18 19:36:29.000000000","updated":"2022-04-14 02:55:29.000000000","submitted":"2022-04-14 02:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825163-tempest-damian-2021-12","meta_rev_id":"cba520ca0451fcdbc94ee9b760f6f201d964c2d5","_number":825163,"virtual_id_number":825163,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 02:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2022-01-29 10:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-31 14:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":2,"date":"2022-04-13 11:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":1,"date":"2022-04-13 11:38:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 22:55:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-29 10:25:09.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-01-31 14:24:36.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2022-04-13 11:38:05.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c4f26d5197a7f21572ed3013d61e84994718a839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-01-18 19:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dff7e4469a9401557cb6539a0aec92de8634b64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 22:55:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15d854c3abc44222bcec622cf7cfa9c9 : SUCCESS in 13m 16s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e60b580a23f94b63aff9f9d02f724ce1 : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/923041903dff469898eba57e9b0ff03c : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8aae4071b434deaaa6b3aaaae5cc86d : SUCCESS in 1h 21m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bb556005732455ea4e3b41cecb1713d : SUCCESS in 1h 33m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b858a70890546a1acdcad41d096e24e : SUCCESS in 2h 18m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a94108a40daf46c985a5b1276c453fdf : FAILURE in 29m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cf734ef0c754ba682ce12118865f34a : SUCCESS in 59m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebb5315d4f24ef8878e4c8231d7c58e : SUCCESS in 7m 46s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/13c8f1c9c8af480c9c54a32247f98973 : FAILURE in 33m 33s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4516434340c14b85a000f9bb4e71605b : SUCCESS in 3h 10m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0ac41eeaba214962b68e3b6b9d47bcd0 : SUCCESS in 1h 36m 36s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/6cbcff35b7714efab986e407c10e724f : SUCCESS in 3h 10m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5a055f271f3c49d196d0e6de1075a159 : FAILURE in 5m 05s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d621f953b4404a578a8bcf02f5a221b5 : SUCCESS in 1h 39m 01s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/184c669fcfcb4d26ade30eecc8c26a91 : SUCCESS in 1h 25m 42s","accounts_in_message":[],"_revision_number":1},{"id":"7e557eb2a740cef7f56e9af272e9eedabec9d6d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-01-18 23:12:22.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7d985e2883261539bd872b562acf4c7f811adc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 02:58:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/140682abc549457aaf36132b0ca358d3 : SUCCESS in 15m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a8363e28e7e146ef8b9ecefbe8f91251 : SUCCESS in 1h 10m 06s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c2eab2eeaef4c02b7a3d1ad8ff0ec3d : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56880b5910a447699eb325ce0d459006 : SUCCESS in 1h 23m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8fd2c36b2a624a3dbfa79b0a2c9a8f8b : SUCCESS in 1h 55m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d422ddc734c49578deb3046f446bded : SUCCESS in 2h 08m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/335edc0f8d8c4881aff436e677f70b6e : FAILURE in 28m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78920039483b43d1becc8c2272ae953a : SUCCESS in 56m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e97fab8413345119ad4799053ac3582 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/98b875b7f9e641d6834ed31c0f2b0909 : FAILURE in 30m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f730723059e742329ceca35d1072add8 : SUCCESS in 3h 34m 04s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e658208f01ab41818f46b69e108a6bd9 : SUCCESS in 1h 24m 52s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/d6c0855c40684467a7b680d9bb80f13a : SUCCESS in 3h 12m 49s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7bfbf68331f94140a2310e891e9a007a : FAILURE in 4m 55s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f3df2eb188654c1293bca0a7d3cca1e2 : SUCCESS in 2h 02m 54s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/c7021d2c37cf4de5bb6b9c51064fbd89 : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":2},{"id":"68031a90d62f6e2eaba2dc2ff9282f636c88c078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-01-19 17:27:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a96a6be8ff3b02909be74294de28f31e3a7471ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 17:29:18.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"3b865552f5996e4023b3479ef0f2e846f30cfdaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-01-19 17:35:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6bb4497cdaebfb89990f5ac13153d80fd45d1b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 20:55:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12d6f27560a5410cb55523c8cf49dcf2 : SUCCESS in 18m 24s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f292945481d84c528f541fb228acfd54 : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c722b003c614b6ba449b80cfbc2cd2a : SUCCESS in 1h 28m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98b07e57a81543618286624e17874db2 : SUCCESS in 1h 29m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/343882dc4abd428eab11b561f7f03a5b : SUCCESS in 2h 03m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0d823c94b1648ddb5920a1390bfa6ac : SUCCESS in 2h 07m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/43222b21676542ad9f10a5bdb750a45e : FAILURE in 33m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7149563bdb2e4a84847048bb3a0a8045 : SUCCESS in 57m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0e602897584dc89949b0bdadb1bff6 : SUCCESS in 6m 59s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d3b39a9b51a45e7ac6eaa32e2cc1432 : FAILURE in 29m 46s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9dcfafd697e943678c8553b1ca7ec94a : SUCCESS in 2h 40m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5c21cc114a584e0985ff02a95c234deb : SUCCESS in 1h 14m 07s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/5bf57838ca174b73b231fcafe2063fda : SUCCESS in 3h 12m 04s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c65306682cc94e9dae52a387d29ecac6 : FAILURE in 5m 19s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/64d97b38f7974ff39e1d41cbba40bd7f : SUCCESS in 1h 19m 57s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/4024883d981440c5b0a89bf16c9f43b0 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":4},{"id":"80020cf35a21518b322092b23ab3aa0bcda3ad07","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-29 10:25:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"73bd167dfbdcab452a02fed598efa97844af8b75","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-01-31 14:24:36.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"eac860f9c19cd23c0411a7db9daa21ec041d2c6f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-13 11:38:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8fafb24ef6f4603b89371ecd2bdb893867f099c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 11:40:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23a06fb5c2654dcc47b315d36b0b646ce3aca1a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 13:46:52.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"f7d336d7a0695916f9eb25890e5cde38e4901304","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-13 21:31:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"94f9a3c1086347be2ec81d2e44157cfd694b2c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 00:06:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1011ea4edcd4e5b8a08504eab652f3e : SUCCESS in 16m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c666bbdea52e41499280757273d8121d : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9e95851af7148bc86493c37eb5b0504 : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bca685fcf14143fe8ffc6e9ed16e44f4 : SUCCESS in 1h 26m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ec8c618b73c54dac96384fead7e283a7 : SUCCESS in 1h 20m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa7920518c2444508f3502e6e2b12961 : SUCCESS in 2h 14m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61249aba3b254b0f80e1c14cdc5b2dc5 : SUCCESS in 2h 13m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4d2fdf6c6274699a96ab651a4ff2933 : FAILURE in 29m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32ca3a0fc70d4a15855d741e1ab68665 : SUCCESS in 57m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847ecbdfe63a4452a0059b57ca07d10a : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/7adbf982f58f4fc48abe602c9cf65ca7 : FAILURE in 30m 17s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e6ea164d476940ad99110fabc053b5a8 : SUCCESS in 2h 27m 19s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/16259d1a38e940c392b81b6310b5dd7c : FAILURE in 5m 07s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f29b688a1f864802812e0aae053085a7 : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":4},{"id":"8feb6fcb9e7e85f519ce602417078490e5364de2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 00:09:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"33afa652594dcde6f5342845b448357a3b6d478e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 02:52:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d393b457e3e1476597fe29230b63bea4 : SUCCESS in 14m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2ef4735b584f491d9a132cea63a270f7 : SUCCESS in 1h 17m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/976fc7ac84f646848f0f61d45e8e8fb9 : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9324db92bed641fd803ef01befb72d5f : SUCCESS in 1h 28m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/933d688d4a494a07a264c55651b4a413 : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bddb19c8c334868a6f0d72535e3768e : SUCCESS in 1h 03m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02bae8b2757474bb1c0af902b0dbb17 : SUCCESS in 5m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c600c07a571e456c89cf122d47978e19 : SUCCESS in 2h 38m 14s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/69ecf4dfc51c436c83a93c08130b6e1c : SUCCESS in 1h 29m 55s","accounts_in_message":[],"_revision_number":4},{"id":"c2e74787796cf95a342220d3c0737c3fbeb28baf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 02:52:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cba520ca0451fcdbc94ee9b760f6f201d964c2d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 02:55:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1c65f6f1156400b973b4c0a49f13117 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c34cf166de8e28758651a5bbaa8b40150d139440","revisions":{"3895d0519ef40ffe08d8630e34c205725deae19b":{"kind":"REWORK","_number":1,"created":"2022-01-18 19:36:29.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/63/825163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/63/825163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/1"}}},"commit":{"parents":[{"commit":"5c9f4a269c1c6eecb8c601bdc94c58eed7754298","subject":"Implement variable: tempest_endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5c9f4a269c1c6eecb8c601bdc94c58eed7754298"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-18 19:01:35.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-18 19:01:35.000000000","tz":60},"subject":"Rename [orchestration] section to [heat_plugin] in tempest.conf","message":"Rename [orchestration] section to [heat_plugin] in tempest.conf\n\nheat-tempest-plugin uses the [heat_plugin] section in tempest.conf\n\nhttps: //opendev.org/openstack/heat-tempest-plugin\nChange-Id: I3f50443b94a5c0e1ebcc566cd268af58561c5e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3895d0519ef40ffe08d8630e34c205725deae19b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3895d0519ef40ffe08d8630e34c205725deae19b"}]},"branch":"refs/heads/master"},"5c469f9a9593e03b4b58219767ff0bdc85d9e51f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-18 23:12:22.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/63/825163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/63/825163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/2"}}},"commit":{"parents":[{"commit":"3122d81219958e34c3b52d8ae452e8f53da13a5a","subject":"Implement variable: tempest_endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3122d81219958e34c3b52d8ae452e8f53da13a5a"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-18 19:01:35.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2022-01-18 23:12:22.000000000","tz":0},"subject":"Rename [orchestration] section to [heat_plugin] in tempest.conf","message":"Rename [orchestration] section to [heat_plugin] in tempest.conf\n\nheat-tempest-plugin uses the [heat_plugin] section in tempest.conf\n\nhttps: //opendev.org/openstack/heat-tempest-plugin\nChange-Id: I3f50443b94a5c0e1ebcc566cd268af58561c5e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5c469f9a9593e03b4b58219767ff0bdc85d9e51f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5c469f9a9593e03b4b58219767ff0bdc85d9e51f"}]},"branch":"refs/heads/master","description":"Rebase"},"1c595e6e83eb40b0477265e72613eb48b9000227":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-19 17:27:53.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/63/825163/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/63/825163/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/3"}}},"commit":{"parents":[{"commit":"b908812a476bc559a725ec89bf49f7afaa15a369","subject":"Implement variable: tempest_endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b908812a476bc559a725ec89bf49f7afaa15a369"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-18 19:01:35.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2022-01-19 17:27:53.000000000","tz":0},"subject":"Rename [orchestration] section to [heat_plugin] in tempest.conf","message":"Rename [orchestration] section to [heat_plugin] in tempest.conf\n\nheat-tempest-plugin uses the [heat_plugin] section in tempest.conf\n\nhttps: //opendev.org/openstack/heat-tempest-plugin\nChange-Id: I3f50443b94a5c0e1ebcc566cd268af58561c5e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/1c595e6e83eb40b0477265e72613eb48b9000227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/1c595e6e83eb40b0477265e72613eb48b9000227"}]},"branch":"refs/heads/master","description":"Rebase"},"c34cf166de8e28758651a5bbaa8b40150d139440":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-19 17:35:08.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/63/825163/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/63/825163/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/63/825163/4"}}},"commit":{"parents":[{"commit":"23387a5fbde05721f2342a74d12a88da6ebdcfb7","subject":"Implement variable: tempest_endpoint_type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/23387a5fbde05721f2342a74d12a88da6ebdcfb7"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-18 19:01:35.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2022-01-19 17:35:08.000000000","tz":0},"subject":"Rename [orchestration] section to [heat_plugin] in tempest.conf","message":"Rename [orchestration] section to [heat_plugin] in tempest.conf\n\nheat-tempest-plugin uses the [heat_plugin] section in tempest.conf\n\nhttps: //opendev.org/openstack/heat-tempest-plugin\nChange-Id: I3f50443b94a5c0e1ebcc566cd268af58561c5e03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c34cf166de8e28758651a5bbaa8b40150d139440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/c34cf166de8e28758651a5bbaa8b40150d139440"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
