)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~825405","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I969ba3f83c053ec22ec02ed58a97b71a3f9686b3","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"tempest-damian-2021-12","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-04-27 14:47:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I969ba3f83c053ec22ec02ed58a97b71a3f9686b3","subject":"Remove unused variables","status":"MERGED","created":"2022-01-19 22:46:57.000000000","updated":"2022-04-27 14:47:37.000000000","submitted":"2022-04-27 14:47:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"825405-tempest-damian-2021-12","meta_rev_id":"2b78fce8bba8670f375c91b0152f973904a70b1f","_number":825405,"virtual_id_number":825405,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 14:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-04-27 14:47:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-27 11:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 11:41:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-20 01:58:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 15:58:06.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-04-20 21:37:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0276e8a96b268ded42574fa14d9a4b008d8379c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-01-19 22:46:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e9c8871e83335d20b324414831cd1038f5dd338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 01:58:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53fb4fdce091442dbe9a64600ce17c47 : SUCCESS in 17m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2af4273048194e05abc07876974962ca : SUCCESS in 1h 21m 09s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/97c869f2c33e4d2b92a82dd865ffa5a6 : SUCCESS in 1h 29m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba5c437fef294559aa08d963cb8027f0 : SUCCESS in 1h 35m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8122f51edc394511b17ae9f74f3de97f : SUCCESS in 1h 53m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/915aa16e87884e81bf0c719146592678 : SUCCESS in 1h 26m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b2992949bb3477395d4a5d2049471d1 : FAILURE in 35m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4de46adf1c4541f992840d50815ee0b8 : SUCCESS in 1h 03m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be9b93ec2fe14e4d9f7c1bef00a428e0 : SUCCESS in 5m 57s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d8ce9528509449fbac157fc2907ca94 : FAILURE in 29m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ab1ce3e7a6434ed8929ec0719e135a81 : SUCCESS in 2h 12m 28s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e582587a4a7a4ac897e548d0d1cd3261 : SUCCESS in 1h 34m 37s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/b525f8bfb23442d9bd5a55740b98bf21 : SUCCESS in 3h 04m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b8f08b7b251645028cf7ca65d7c21565 : FAILURE in 4m 49s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/86f16652e1e24be78e67af91bb349dd5 : SUCCESS in 1h 09m 05s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/06558c96a4774231b1e321a07072599c : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7f430890ded53f36b3d7edd9f289a34b97cbbaf2","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-19 15:58:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d461e7002f5500fc71aae0384198112163044ceb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 21:37:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d28f240e123b5d6bbf3dd23a20f448b56374a271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-04-26 16:40:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"68f3dfab51a2f82ef1afe0bc799e61b69ed4bbf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 16:41:43.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible-os_tempest for 803477,9","accounts_in_message":[],"_revision_number":2},{"id":"f386510594d53aa5ad58aaac2385ff7a61174db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-04-26 16:45:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"95d0804c0ae1c1537adb48ff75328c9e17a1f26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 19:15:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24f39beceb564b82bdaf47cdd01eb8bc : SUCCESS in 14m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0c48c0e4aedc47ffa8e4dc4ae706375c : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5749ac0756084d459510922e63ae3be2 : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e2326a4aaa24e7d8fe3997dd4c4e6a2 : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/fc0e2fc5de3a49d7be723549927f5be9 : SUCCESS in 1h 30m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/81524df27836451aaeff044ba0642e03 : SUCCESS in 2h 02m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27fb1b7147b94969ab7de163e47b7ff5 : SUCCESS in 2h 02m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2b89d02acd148b0badc2b9b03271d86 : FAILURE in 31m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4623507c72574830be175f291f7c7a26 : SUCCESS in 1h 02m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376699b9629c426fa263933da208bc6c : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b33342e9aaa4260bf71124eb1c5d795 : FAILURE in 32m 31s (non-voting)\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1b4bb67b9df649c1a3c9203ffb42203a : SUCCESS in 2h 19m 59s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2ea4bd35bc8f46de862f0c9a9c9d5030 : FAILURE in 4m 51s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8d2be0433bd04cf78757ec212c9bd9ab : SUCCESS in 1h 14m 05s","accounts_in_message":[],"_revision_number":3},{"id":"deeaf937d5d506db22d54612eb7aee1601ce74fd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 11:41:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"871d8ba1e3bf25b04ccb9a8238b33441d766d9ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 11:45:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b719715c13bb569ae626ca4666fd7ffdedfb939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:47:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaa10b24ea4e405fb162ae07a6bd795c : SUCCESS in 13m 09s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ddc9fdc7f8034cadb590f00fb5b58bf0 : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ea7af49009447f4886abe4ea6f3acc7 : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c42b6ec009424e4bbaaa7aeb69a00468 : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/74ec83496e664750bf6ff47a6db18028 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6c76253ee75437899d0fed6fec65475 : SUCCESS in 1h 01m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92f0a85de1e4431af38d7aa13ad9d94 : SUCCESS in 6m 25s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5ad69f3d481442feb387ef6301085b38 : SUCCESS in 2h 03m 25s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/54dd13cccbeb4754a1b799544f3b106c : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":3},{"id":"212bfe33ed4fd710b3cd3dce5ce342fed6e2d53a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:47:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d194e546f6a32c715aa33487beb8eee129355207","revisions":{"2fccbc172c1352e82c8f97a302228a4efd8ea710":{"kind":"REWORK","_number":1,"created":"2022-01-19 22:46:57.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/05/825405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/05/825405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/1"}}},"commit":{"parents":[{"commit":"12baace6702d4d954da5c935ad39be2351f511dc","subject":"Add support for both Credential Provider Mechanisms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/12baace6702d4d954da5c935ad39be2351f511dc"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-19 22:45:47.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-19 22:45:47.000000000","tz":60},"subject":"Remove unused variables","message":"Remove unused variables\n\nFew variables defined in defaults/main.yml are not used anywhere.\nWe should remove them.\n\nChange-Id: I969ba3f83c053ec22ec02ed58a97b71a3f9686b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2fccbc172c1352e82c8f97a302228a4efd8ea710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2fccbc172c1352e82c8f97a302228a4efd8ea710"}]},"branch":"refs/heads/master"},"d3f88a5716dd886b93188470de6dcc6d9e3fbb25":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-26 16:40:47.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/05/825405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/05/825405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/2"}}},"commit":{"parents":[{"commit":"81034c7722176c807f1f33bc2d5d7661a93909a8","subject":"Add support for both Credential Provider Mechanisms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/81034c7722176c807f1f33bc2d5d7661a93909a8"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-19 22:45:47.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2022-04-26 16:40:47.000000000","tz":0},"subject":"Remove unused variables","message":"Remove unused variables\n\nFew variables defined in defaults/main.yml are not used anywhere.\nWe should remove them.\n\nChange-Id: I969ba3f83c053ec22ec02ed58a97b71a3f9686b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/d3f88a5716dd886b93188470de6dcc6d9e3fbb25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/d3f88a5716dd886b93188470de6dcc6d9e3fbb25"}]},"branch":"refs/heads/master","description":"Rebase"},"d194e546f6a32c715aa33487beb8eee129355207":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-26 16:45:21.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/05/825405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/05/825405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/05/825405/3"}}},"commit":{"parents":[{"commit":"0566c20d99d2b6b89d1b6b6b0aed067b43487e38","subject":"Add support for both Credential Provider Mechanisms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0566c20d99d2b6b89d1b6b6b0aed067b43487e38"}]}],"author":{"name":"Damian Dabrowski","email":"damian@dabrowski.cloud","date":"2022-01-19 22:45:47.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2022-04-26 16:45:21.000000000","tz":0},"subject":"Remove unused variables","message":"Remove unused variables\n\nFew variables defined in defaults/main.yml are not used anywhere.\nWe should remove them.\n\nChange-Id: I969ba3f83c053ec22ec02ed58a97b71a3f9686b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/d194e546f6a32c715aa33487beb8eee129355207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/d194e546f6a32c715aa33487beb8eee129355207"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
