)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~826121","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I83fa8c729967046829db638500b65faa7f55a50d","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"826121","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-01-26 10:55:23.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"hashtags":[],"change_id":"I83fa8c729967046829db638500b65faa7f55a50d","subject":"Define variables for tempest plugins","status":"MERGED","created":"2022-01-24 16:19:18.000000000","updated":"2022-01-26 17:23:26.000000000","submitted":"2022-01-26 17:19:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826121-826121","meta_rev_id":"f5e085548e4c6e9186025f07561893107e1bde96","_number":826121,"virtual_id_number":826121,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 17:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-26 08:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-25 22:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2022-01-26 08:13:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-24 19:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 22:30:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-01-26 08:13:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"0d454b13b98be0ee5713e323a3ddba1563561a03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-24 16:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efbcd6c946eab7f61a41c86b07317522ea10c049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 19:40:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/943f458f63924fbeb0f1c12a742f4931 : SUCCESS in 13m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3ae936108f4a4325918df457a4840417 : FAILURE in 1h 10m 12s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6047f1bae66d42b3b7d02f2a1a372562 : FAILURE in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28c72ca53c7540d494b88b739102980c : FAILURE in 1h 18m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8bf2eb7afb1a4847a6aad0ffdc038ebf : FAILURE in 1h 46m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/543b144defd34b63b7c149dc5fb22512 : FAILURE in 1h 43m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/75ef0b7b98284f04b6fa2667f4cb9e2d : FAILURE in 29m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fe973ec059746a584eb8850e58e6226 : FAILURE in 45m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17aec47062d04c7397da05e69eba4e5b : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b09883b962546fdace783e0e15350f4 : FAILURE in 28m 37s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/710186df4c1e42ef839a41bd5c877ccc : SUCCESS in 3h 02m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2a5f455942ec4af5bffe9cccdf28a1df : SUCCESS in 1h 16m 08s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/0e38d99a5ccb47deab985aef164ba6dc : SUCCESS in 3h 04m 47s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/96c77ede7caa44dd861cba4060c5faa0 : FAILURE in 4m 41s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7c736397f1b04632a1c7b783b5b81d67 : SUCCESS in 1h 40m 35s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/38d8941e69a14fa3935d00426414009d : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e965e55446211657398ab9a5f38083f2c9a9d91e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-25 15:36:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"69cbb4544edac13ea7c0b567a22f09c7ee745bb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-25 15:39:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec3662b1ba8510880ff8a429ef7f7cd0e1c45ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-25 18:10:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2f6bcde14ffc2747ced1d180b2c8ab3a6b36bb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 21:48:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5b3edf698254058b78504a7ffb7b9a0 : SUCCESS in 19m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3bfcfe1a23aa48d4aebc823f4fef73d4 : SUCCESS in 1h 18m 47s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b90146cd4bb74f3593cad79064758cc2 : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a24e7fab423344baad2acb1b7e8eb545 : SUCCESS in 1h 30m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a343bc661444d4b98d0f6c57ae7c09f : SUCCESS in 2h 04m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e5135e2bf87485aa85e0aee43b7bfdb : SUCCESS in 2h 10m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b8a05c00d9c4631985e5566a826d7d4 : FAILURE in 33m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca8bbdace64941b49f5551398ef0c194 : SUCCESS in 58m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef5514c1e8c4d31a26dabce8794f19c : SUCCESS in 6m 00s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e1e2d6dbd394e1ba686983b6dd076ce : FAILURE in 35m 27s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/dc578788d5ca45958493bfcf23f81e9b : SUCCESS in 2h 52m 40s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/9526fb41af12437ab3e08457d7e71521 : SUCCESS in 1h 45m 31s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/0832e390d9624bd882fd5dbf1c8f9c04 : SUCCESS in 3h 25m 37s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/23b9855e739d41a7a039bd6878e259fb : FAILURE in 5m 33s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0d65e40ce43f4c3ba6278a2aa37d2321 : SUCCESS in 1h 27m 18s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/9bd7f47e16dd40c09ced99cc116c541d : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":4},{"id":"ec48c214d6aafab9d4c0193ef4b1f6c20c621b24","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-25 22:30:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"84a3194531649b1332e9874a4ef567551f94a3ee","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-26 08:13:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"075eb9b7d8cc01654699e68b8caf80073f206b3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 08:15:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e5b42fd6829fb0267ed3e60f316abd4da54f7521","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 10:50:42.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921d911872254e4eba2c58e9f849b1f7 : SUCCESS in 14m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d71cdadda2eb405287be9387aa5374b1 : SUCCESS in 1h 01m 31s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/298feca7f8764d29b19ca709a171ae5e : SUCCESS in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bf68b5aa47a41f5a61458ee87524116 : FAILURE in 1h 02m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f77f85ad6cb43b49a841d1c804ecf7e : SUCCESS in 1h 02m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a004ab1ff29d42389bc3844864f54bd8 : SUCCESS in 5m 38s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/10f495e1591f4fb9b99179c64a4cc1ae : SUCCESS in 2h 33m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ed52ca90d0e34e1d8668373db0e6ad39 : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":4},{"id":"2f32ec4b236ecde76b2e5bad81c6bd90fba55381","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-26 10:55:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"23ebfc44eff8c5130435544b9b93c9389384ce30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:04:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/230d2f219278432593df15db8ead04c6 : SUCCESS in 13m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/624cc9d3888c4e0aa0f5649ed3b9b90a : SUCCESS in 1h 07m 49s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b64bbf7907cd435e9de9c9297d27ac06 : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5eecf37c504b17b743edf5b3a8785e : SUCCESS in 1h 33m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b82a700878e14b64bb185d543bd143f2 : SUCCESS in 2h 12m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff88590ab0e2488c95d982c4e9b238da : SUCCESS in 2h 19m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5b55960c4484fd28e1b13212761dc79 : FAILURE in 31m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0a3013fed7444bba3028a538467442 : SUCCESS in 59m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4254ed15fb824d90a4ea8387516d09b8 : SUCCESS in 5m 54s\n- openstack-ansible-deploy-aio_distro_metal_tempestconf-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9238f0a5d774e499bf9906ca69aaf5f : FAILURE in 30m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/de6e79181c8e4392ac1f2693b2fc31ae : SUCCESS in 2h 54m 51s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7cb90afbb7934d3b8fd10cbe2764c6b9 : SUCCESS in 1h 19m 23s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/a4308bbfecbf48eb8bc8ad6eff0ea92d : SUCCESS in 3h 02m 33s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f2336a02b42a4c409320723d15ef5957 : FAILURE in 5m 25s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a72d32b251fa42a0b834ca33bc41ce63 : SUCCESS in 1h 26m 36s\n- tripleo-ci-centos-7-standalone-os-tempest-train https://zuul.opendev.org/t/openstack/build/640454e4d3bb4d84ab9fd06dc2e61db2 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":4},{"id":"43b75491a2edf9c7a696f4401aa3d305c60b5ef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:07:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ca8078a46b525dc153f99668313d7ced46e71391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 17:19:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0588110057c04233b63d73d1b67cd204 : SUCCESS in 14m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b77ccb96b114b54a9d43076b4baefca : SUCCESS in 1h 05m 27s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f90ba924008648c0bc1290cc6500ee3e : SUCCESS in 1h 21m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1213e0f2b37b4eb2b73c46868b40d046 : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f5da8cc6fd946989f3562fb40ae87c9 : SUCCESS in 58m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c985b3723e43ed93ef941cb105a72c : SUCCESS in 6m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/118ea31ffff541b78ad3c5697ad357c5 : SUCCESS in 3h 07m 38s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4a7375866f6f4d37a961a60b10918554 : SUCCESS in 1h 43m 02s","accounts_in_message":[],"_revision_number":4},{"id":"96c78196f3721dc460def575bdf9abd53a849aca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 17:19:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f5e085548e4c6e9186025f07561893107e1bde96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 17:23:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce794b94013461fa20a5323b9d5d4b9 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6f75a079827068a15d6dc378ab3ecd0133da4a46","revisions":{"0f677f6b50186236c84ce2221dcf09072b658e82":{"kind":"REWORK","_number":1,"created":"2022-01-24 16:19:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/21/826121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/21/826121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/1"}}},"commit":{"parents":[{"commit":"0bb6f1f32e2a9dcc00d8382b8413dd68b3df8bcb","subject":"Deprecate congress and lxd tempest plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0bb6f1f32e2a9dcc00d8382b8413dd68b3df8bcb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-24 16:18:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-24 16:18:41.000000000","tz":120},"subject":"Define variables for tempest plugins","message":"Define variables for tempest plugins\n\nThis allow us to define and later bump specific version of tempest\nplugins for stable branches and override source of the repositories for\ndeployers.\n\nChange-Id: I83fa8c729967046829db638500b65faa7f55a50d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0f677f6b50186236c84ce2221dcf09072b658e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0f677f6b50186236c84ce2221dcf09072b658e82"}]},"branch":"refs/heads/master"},"77167e70aacacff88ce7bc98f065ded826e39112":{"kind":"REWORK","_number":2,"created":"2022-01-25 15:36:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/21/826121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/21/826121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/2"}}},"commit":{"parents":[{"commit":"0bb6f1f32e2a9dcc00d8382b8413dd68b3df8bcb","subject":"Deprecate congress and lxd tempest plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0bb6f1f32e2a9dcc00d8382b8413dd68b3df8bcb"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-24 16:18:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-25 15:36:26.000000000","tz":0},"subject":"Define variables for tempest plugins","message":"Define variables for tempest plugins\n\nThis allow us to define and later bump specific version of tempest\nplugins for stable branches and override source of the repositories for\ndeployers.\n\nChange-Id: I83fa8c729967046829db638500b65faa7f55a50d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/77167e70aacacff88ce7bc98f065ded826e39112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/77167e70aacacff88ce7bc98f065ded826e39112"}]},"branch":"refs/heads/master"},"3eef230413aa3d155b78cd05e2b33de511ff6c78":{"kind":"REWORK","_number":3,"created":"2022-01-25 15:39:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/21/826121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/21/826121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/3"}}},"commit":{"parents":[{"commit":"1900f0ff6bbef1651cc502e9e880244b19656554","subject":"Deprecate congress and lxd tempest plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/1900f0ff6bbef1651cc502e9e880244b19656554"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-24 16:18:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-25 15:39:24.000000000","tz":120},"subject":"Define variables for tempest plugins","message":"Define variables for tempest plugins\n\nThis allow us to define and later bump specific version of tempest\nplugins for stable branches and override source of the repositories for\ndeployers.\n\nChange-Id: I83fa8c729967046829db638500b65faa7f55a50d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3eef230413aa3d155b78cd05e2b33de511ff6c78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/3eef230413aa3d155b78cd05e2b33de511ff6c78"}]},"branch":"refs/heads/master"},"6f75a079827068a15d6dc378ab3ecd0133da4a46":{"kind":"REWORK","_number":4,"created":"2022-01-25 18:10:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/21/826121/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/21/826121/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/21/826121/4"}}},"commit":{"parents":[{"commit":"1900f0ff6bbef1651cc502e9e880244b19656554","subject":"Deprecate congress and lxd tempest plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/1900f0ff6bbef1651cc502e9e880244b19656554"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-24 16:18:41.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-01-25 18:10:46.000000000","tz":0},"subject":"Define variables for tempest plugins","message":"Define variables for tempest plugins\n\nThis allow us to define and later bump specific version of tempest\nplugins for stable branches and override source of the repositories for\ndeployers.\n\nChange-Id: I83fa8c729967046829db638500b65faa7f55a50d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f75a079827068a15d6dc378ab3ecd0133da4a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6f75a079827068a15d6dc378ab3ecd0133da4a46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
