)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~857674","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~Ibffaadb721796d6f657972633d19bdd36e411c52","project":"openstack/openstack-ansible-os_tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2022-09-20 19:50:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibffaadb721796d6f657972633d19bdd36e411c52","subject":"Remove redundant vars line","status":"MERGED","created":"2022-09-14 12:58:13.000000000","updated":"2022-09-20 19:51:41.000000000","submitted":"2022-09-20 19:50:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857674","meta_rev_id":"3fec0b617d7b359e19fea33ad4afa4bc1b19658d","_number":857674,"virtual_id_number":857674,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 19:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-09-20 17:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-09-16 13:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-09-20 17:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 15:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 13:41:35.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2022-09-20 17:27:47.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"8d183bc2c9f418cc6ff2a2da253491677af6e8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2022-09-14 12:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"959001f96ab65bfa93972e08528cd3d423e89365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:43:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6805922dfc41495f8aece658efaf8ef3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9cb60ee5f0a434287c50b729042121c : SUCCESS in 37m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/90a7d5a763eb42a7a076bba34c204a46 : SUCCESS in 1h 40m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b3d603744cb42ce98f35e137f267461 : SUCCESS in 1h 50m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dc7a9069e804cd9b64727066aeb6bcd : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/031c1a7f66ed419b9a9bb26cd4d9c8ab : SUCCESS in 1h 16m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b837d114d29b404daea33f06cd5f1fea : SUCCESS in 1h 47m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0199d471b292458eb4aeaa2f5f1ec273 : SUCCESS in 1h 52m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f58c64bee583496b843faa415d39406f : FAILURE in 29m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f4281985b324af0bfb066e77dc31fd3 : FAILURE in 29m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d568a4f8fca42a095ebe1dbf54d10d5 : SUCCESS in 6m 26s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/130c8aba4eac4d28bbca7feabd69154a : SUCCESS in 2h 38m 04s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/25864d420f7c496b92adc95f86e81ea6 : FAILURE in 5m 19s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/97a055790f5b4cc780fc2588266353fc : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a2c6fc41dc242ea2fa529348f83312fb74a54eb0","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2022-09-16 13:41:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02f84177909cc69788ee89dcc86c0415b655d4e8","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-09-20 17:27:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1579805653b5cec56a38d27042c6cfaa53669585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 17:31:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"befce49fae28ff003331504e216e31a8c515c1ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:50:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63e2b152bab24a91b58887c23f55063a\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b16ef96670344e0583e1e0b107897080 : SUCCESS in 19m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a0ea9741939d40c6be13a762a321a63f : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/19d873f5bb7143c595ef8f029e5fc729 : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59931ff35ba54ec1bfe5412a36ef2eba : SUCCESS in 1h 31m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c0bf5bdde41a98aba0896cf618d9e : SUCCESS in 5m 37s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/99256418df6b40cdabf5f3ba656d43d8 : SUCCESS in 2h 18m 29s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/15de32fa062940879f1b0dc41ee997a9 : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c10889e5a9685c5cfc06a5848b61f9c37d1e7ec0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:50:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3fec0b617d7b359e19fea33ad4afa4bc1b19658d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 19:51:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0eb7d67acf04716adfe4aea26c58018\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330a1b8ac0d046ca957fdbbe1c642344 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a9d57f637c856fffe04fe52f976afc3ab3e334f","revisions":{"2a9d57f637c856fffe04fe52f976afc3ab3e334f":{"kind":"REWORK","_number":1,"created":"2022-09-14 12:58:13.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/74/857674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/74/857674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/74/857674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/74/857674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/74/857674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/74/857674/1"}}},"commit":{"parents":[{"commit":"bb33438d517eceb4bf95da7b999475b0d7ed5681","subject":"Fix dynamic service enablement for octavia and magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bb33438d517eceb4bf95da7b999475b0d7ed5681"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2022-09-14 12:53:52.000000000","tz":120},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2022-09-14 12:58:11.000000000","tz":120},"subject":"Remove redundant vars line","message":"Remove redundant vars line\n\nThis line was introduced by I0230f1a93d16243fc50df79b2d0780e744706c7b\nand should already be covered by the distribution_major_version line\nabove.\n\nChange-Id: Ibffaadb721796d6f657972633d19bdd36e411c52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2a9d57f637c856fffe04fe52f976afc3ab3e334f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2a9d57f637c856fffe04fe52f976afc3ab3e334f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
