)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~858260","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I949f522348af95b957df4682fe342cb245382dad","project":"openstack/openstack-ansible-os_tempest","branch":"master","topic":"1990045","attention_set":{},"removed_from_attention_set":{"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-09-20 14:49:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I949f522348af95b957df4682fe342cb245382dad","subject":"Reordering variable definitions to prevent empty includelist edge case","status":"ABANDONED","created":"2022-09-19 08:19:03.000000000","updated":"2022-09-20 14:49:46.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"f143d285bd3d4af2ce2280fcf627b0a6da6c5886","_number":858260,"virtual_id_number":858260,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-20 13:58:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"reviewer_updates":[{"updated":"2022-09-19 10:59:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 11:20:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"ea47b5a48fb99450a1037675b8c4a59cb6f1e5e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-09-19 08:19:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5e21f2c3d5a00c796ba83f38bcab42170f2184b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 10:59:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbf0bd07527492c98201069b64b5a3d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8406605544bb42848c2a8400b405713e : SUCCESS in 11m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dd3fdf64b3fb445887215316d4c757a7 : SUCCESS in 1h 11m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/049d35bfef7148d899cca651612d7e96 : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8456976c1e147c6bd6e4a58c2c98765 : SUCCESS in 1h 22m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a73093048144d0e84805882b45b4a9a : SUCCESS in 1h 11m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec8995088e254c79b3eca08925fa6e06 : SUCCESS in 1h 54m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bfdbba047824be688bcd826b82ff3e6 : SUCCESS in 2h 02m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9607286da144aa3be3313071630f22d : FAILURE in 29m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dba701d7c889474db47a0429632369e0 : FAILURE in 27m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d5d851d1e649b9b8c7d53c3736ce21 : SUCCESS in 4m 34s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/baff8afa5709487bb02cfdfb8db2f2d5 : SUCCESS in 2h 33m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/dd0cb2e7637348758d236ea60c98e2dc : FAILURE in 3m 38s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/afeb28bcf4c94bd49af4ac8fed6282af : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"52b269b83effd1465fa0cc11f857d3ee223222c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-09-20 11:16:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"18cc3ecfffcfd9cbc55b3d3e3a2b2b83ca5b5735","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-09-20 11:20:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39f4653396bce72d02913fa96c08dd46f89d4ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 13:58:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/970e74a4e2714a1f853b8b9219fb4074\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88189dc5af314f43ae1c24ffdf6642ac : SUCCESS in 21m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6d3baf93647443469ebb9ebd3178e8eb : SUCCESS in 1h 27m 02s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/44e514fd317d4f02ab84575c84db8505 : SUCCESS in 1h 28m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b67446e909c47dabbb0abea760efbb1 : SUCCESS in 1h 33m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab76f4d90da04f9a9bbb937393941921 : SUCCESS in 1h 25m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffa21bdb5f2347cbbdc2b312f4168742 : SUCCESS in 1h 59m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/649be57a636e44888943019ef78af045 : SUCCESS in 2h 05m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/68f7eca23cd34635ae28661c77995385 : FAILURE in 35m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4fb7eed0199480b868c1cb13ebad4ac : FAILURE in 35m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd09f45b7874c2a9cb35dfbd3d4df91 : SUCCESS in 6m 30s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/f9e6efeefdfb451ca08270175a45c4d4 : SUCCESS in 2h 35m 39s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c22294a7e1c64c5fb27ec10e25f00c25 : FAILURE in 4m 41s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/96a34d5b10d4437ba7b828fb576bfd5e : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":2},{"id":"f143d285bd3d4af2ce2280fcf627b0a6da6c5886","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-09-20 14:49:46.000000000","message":"Abandoned\n\nUnnecessary, ansible variables are resolved at runtime.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad4d679b41757b08852f33f054f5376ae5008ab8","revisions":{"b4bc48101d8c626e6d9dbd9838ddfe3f6c0ed0db":{"kind":"REWORK","_number":1,"created":"2022-09-19 08:19:03.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/60/858260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/60/858260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/1"}}},"commit":{"parents":[{"commit":"bb33438d517eceb4bf95da7b999475b0d7ed5681","subject":"Fix dynamic service enablement for octavia and magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bb33438d517eceb4bf95da7b999475b0d7ed5681"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-09-19 08:18:50.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-09-19 08:19:00.000000000","tz":120},"subject":"DNM: Testing change for CIX","message":"DNM: Testing change for CIX\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I949f522348af95b957df4682fe342cb245382dad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b4bc48101d8c626e6d9dbd9838ddfe3f6c0ed0db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/b4bc48101d8c626e6d9dbd9838ddfe3f6c0ed0db"}]},"branch":"refs/heads/master"},"ad4d679b41757b08852f33f054f5376ae5008ab8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-20 11:16:01.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/60/858260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/60/858260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/60/858260/2"}}},"commit":{"parents":[{"commit":"bb33438d517eceb4bf95da7b999475b0d7ed5681","subject":"Fix dynamic service enablement for octavia and magnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bb33438d517eceb4bf95da7b999475b0d7ed5681"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-09-19 08:18:50.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-09-20 11:16:01.000000000","tz":0},"subject":"Reordering variable definitions to prevent empty includelist edge case","message":"Reordering variable definitions to prevent empty includelist edge case\n\nThe existing order of variable definitions allowed for scenario where the include list\nwouldn\u0027t have any contents. This in turn would lead to include list file not being created,\npotentially causing job failure as tempest would attempt to read from it.\n\nRelated-Bug: #1990045\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I949f522348af95b957df4682fe342cb245382dad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ad4d679b41757b08852f33f054f5376ae5008ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ad4d679b41757b08852f33f054f5376ae5008ab8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
