)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~883854","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~I5256943b5f175cbf46ae36361887dd055e0c52be","project":"openstack/openstack-ansible-os_tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-05-23 14:51:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5256943b5f175cbf46ae36361887dd055e0c52be","subject":"Make tempest respect service_install_method","status":"MERGED","created":"2023-05-22 15:29:18.000000000","updated":"2023-05-23 14:53:08.000000000","submitted":"2023-05-23 14:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883854","meta_rev_id":"d4f3cb8d7d8eebe316b4dc1e9a588c2dfc5a46a2","_number":883854,"virtual_id_number":883854,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-23 14:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-05-23 11:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-23 12:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-23 12:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-22 17:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-23 11:54:47.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2023-05-23 12:05:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"b5d883664e42b4abee078c474e002de5bb3365a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-22 15:29:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e68dbaada64b5959abb5e5f82a20a96a2a3791c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-22 15:30:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"171fb13ea937b57ba986380b6546aaf65dca1cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 17:53:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79bbdaf4f35f4fd693d06df0e4150ec2\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e023fd0be774f429794277f2a785650 : SUCCESS in 10m 32s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c76339487f8f419395a2b9cf32e06d93 : SUCCESS in 1h 38m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b885aba4e424e188a3ac4989d3c3a91 : SUCCESS in 1h 34m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c75da9165bb4382bfec908c6ac653e1 : FAILURE in 1h 00m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a5b4c422b9846768f06d13785db7226 : SUCCESS in 1h 28m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e290744b1164847a523d0ed0694ef1e : SUCCESS in 1h 36m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/741e078f1a6644ecb9bed8528cef54dc : SUCCESS in 1h 43m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6236b359b10a491ab3dacaa6951208d9 : SUCCESS in 1h 44m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0284cf6557af4f7393fe10ca4e16de0f : SUCCESS in 2h 11m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4f46b61b9ae47528c73e22b4b9d67fc : SUCCESS in 2h 19m 43s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e21ad4607664d529d02d135596f9342 : SUCCESS in 1h 52m 35s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9bb79a2ebb0c47918e907d9010d2c7e5 : SUCCESS in 1h 07m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08a7c1e853d040e498d7130d76c79222 : FAILURE in 43m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53704ec5736b466bb2129fc1348f590f : SUCCESS in 4m 47s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/dd35bab3fcf64852b2897df4d7b6244c : SUCCESS in 1h 51m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/70706197d24740538bdb3452802d33d4 : SUCCESS in 3m 42s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ea917b63b8364dd08a5769f977aa92df : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9daab9c289e5f1132aafd15f361fd95b171175d7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-22 21:01:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5668e26621af0ff49b2dca22a5d6a070bceedbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 23:07:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/629cc197889d41a0a30a3016614613ac\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d0cde4d57654a01a820274b7ab21e48 : SUCCESS in 11m 40s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/97955415b8274efa9aab1982a60c4c8b : SUCCESS in 1h 24m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/64c515c802574ffab1c8ca4e59ad2e6d : SUCCESS in 1h 28m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b29b90d6d3e74988aca41a95885add0e : SUCCESS in 1h 35m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a9e0e9b4cc4c4c81dbebf4c564a6ff : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/febb2a215e0f415ca1eb84d47424e822 : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/576809323d3e4f9b8707354e61b307c3 : FAILURE in 49m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93501f09c34045bb812487b78d324526 : SUCCESS in 1h 32m 23s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4140bd67a3cb481e9ec947a3e90c3d5e : SUCCESS in 1h 47m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0747a4ef05794cc8bb16b8bfe71485b7 : SUCCESS in 1h 57m 26s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3196b8ddb9a4fe3ba0f38185da29929 : SUCCESS in 1h 51m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b4811cceb2a74080a94b7c8d6dc77092 : SUCCESS in 1h 01m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c67f73e4f2c4db495c98c5c95420572 : FAILURE in 46m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb54a5628bc429f8ba2dcca31e52071 : SUCCESS in 5m 16s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c811d7b219a542e3848a41b5b14174ad : SUCCESS in 1h 54m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7682b129fa5f4c83a8d988646350ce74 : SUCCESS in 3m 10s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6cec515bb29a46299eaa7686958db59e : SUCCESS in 1h 03m 18s","accounts_in_message":[],"_revision_number":2},{"id":"12e2b93f9dc3cc3d90086b1c80180de9a3c65fdf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-23 09:05:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18e5153a6c76ee5d1804e0be58eb1c3227430b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 11:25:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97dab199041f456c904b684dc335dc7f\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2650bd06478745d7a37dfaae47f8924e : SUCCESS in 12m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f2fe71f948eb43659dc141ec7a9d99cd : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/752f55f351c44b15b72a7a2bbd3e4450 : SUCCESS in 1h 33m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b4d313976fc4660a6640c2cae3c6335 : SUCCESS in 1h 23m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7bc5eeff4574c42b1c5a7da86f2ad8c : SUCCESS in 1h 32m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52f5dd9a7fd347a3bbe8d4f35d3805d1 : SUCCESS in 1h 27m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/775777238ea94350aaaddd86af005b8e : SUCCESS in 1h 41m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78bc37d1c60148e3845a20f6e5d75108 : SUCCESS in 1h 29m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a98368e6def64012ae71626360e713ea : SUCCESS in 1h 44m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/434b8afc1fa44a9d878dd1d00badb3dc : SUCCESS in 2h 01m 06s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078d95d655d04de7ba378e57b9246e11 : SUCCESS in 1h 58m 11s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4e3031f28c444f188fdc68762318c8be : SUCCESS in 58m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd89fbee5c2f40679c5ee7e7514c8c00 : FAILURE in 49m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5053274d930745578d5aeeedb554a850 : SUCCESS in 4m 49s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8cb829c0f8bc454eb3d170d59b2c91fa : SUCCESS in 2h 00m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/bbf8f86d5c814984bb1144651c19e5e1 : SUCCESS in 3m 10s (non-voting)\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/9f81d5eaa0cd4184b23e23ed0f21c329 : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":2},{"id":"57897447b5f9f6ee737db51f8f11f70cc2ec2471","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-05-23 11:54:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a96fef937bc82304eec0c3e0f18d9f750d5ca7c8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-05-23 12:05:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a28d3b06807707da02085e88cd2dba5f633ecc61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 12:10:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16bb44a0840bc0880394f29c3758134bd55abbe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 14:51:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f1e75d5149e4853adb9074a020a0531\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/682f342c78c2430db3755a5acd894293 : SUCCESS in 9m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/53c59e53cead41e9b03ffdd726755c48 : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/737c0ecfe5fa40a78ee5198e1e62848e : SUCCESS in 1h 31m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/292161d7f25b49e68a74623e49ccdd68 : SUCCESS in 1h 29m 48s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65d107ea51d545a5bf4d9a259e86a9d8 : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/688742c633804bb5a7b596513184c6ef : SUCCESS in 1h 35m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed83e7a5b744a1ea456b22d7275b30d : SUCCESS in 5m 40s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8f253271c859458aa426bb262178e577 : SUCCESS in 2h 06m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/df7e76bcb6df44b5b05ec04f0b1b305c : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":2},{"id":"434f6a253dedaaea497d6be213d2046c82966a22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 14:51:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d4f3cb8d7d8eebe316b4dc1e9a588c2dfc5a46a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 14:53:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ab0ba307fb4d4fa69f74e6a96cc758\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d88dce997ad24496934b60ccfbe447cf : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a8459bec03b433b1487b89368e69fad8d92cfd7","revisions":{"64d2dda7cc99edf2ed3293305bca5b25994cdf7a":{"kind":"REWORK","_number":1,"created":"2023-05-22 15:29:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/883854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/883854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/1"}}},"commit":{"parents":[{"commit":"a871336aa0bfb8c80900e17a78c5770531a52f91","subject":"Handle openstack.cloud.image from collection 1.x or 2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/a871336aa0bfb8c80900e17a78c5770531a52f91"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-22 15:29:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-22 15:29:11.000000000","tz":120},"subject":"Make tempest respect service_install_method","message":"Make tempest respect service_install_method\n\nChange-Id: I5256943b5f175cbf46ae36361887dd055e0c52be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/64d2dda7cc99edf2ed3293305bca5b25994cdf7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/64d2dda7cc99edf2ed3293305bca5b25994cdf7a"}]},"branch":"refs/heads/master"},"0a8459bec03b433b1487b89368e69fad8d92cfd7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-22 15:30:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/883854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/54/883854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/54/883854/2"}}},"commit":{"parents":[{"commit":"a871336aa0bfb8c80900e17a78c5770531a52f91","subject":"Handle openstack.cloud.image from collection 1.x or 2.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/a871336aa0bfb8c80900e17a78c5770531a52f91"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-22 15:29:11.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-22 15:30:38.000000000","tz":0},"subject":"Make tempest respect service_install_method","message":"Make tempest respect service_install_method\n\nFor all services we do respect service_install_method vairable,\nand define \u0027source\u0027 as the default options. At the moment tempest\nattempts to follow source path even for distro deployments.\n\nChange-Id: I5256943b5f175cbf46ae36361887dd055e0c52be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0a8459bec03b433b1487b89368e69fad8d92cfd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0a8459bec03b433b1487b89368e69fad8d92cfd7"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
