)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~886165","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~Ia53e458bf29b0c29480a8036c0387c7ac9997688","project":"openstack/openstack-ansible-os_tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-11 23:44:12.000000000","reason":"removed on reply"},"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-01-12 08:21:09.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-01-23 22:27:11.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-23 22:27:11.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-23 18:17:48.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-01-23 22:27:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia53e458bf29b0c29480a8036c0387c7ac9997688","subject":"Update CirrOS to latest versions","status":"MERGED","created":"2023-06-15 09:08:36.000000000","updated":"2026-01-23 22:29:02.000000000","submitted":"2026-01-23 22:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886165","meta_rev_id":"2519efa86750e90e920b7a16de2971e7f6665ecf","_number":886165,"virtual_id_number":886165,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 22:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-11 23:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-12 08:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-28 16:53:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-01-12 09:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-01-12 09:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 11:39:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-28 16:53:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-28 16:53:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-12-28 16:53:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2025-12-28 16:53:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2025-12-28 16:53:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"54f3e259ccf9371e6f0bef6bcd92ea2a51bfcf0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-06-15 09:08:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d4c035dfc046bfbe2735c1867351cceef27b957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 11:39:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d69ea7471be4817814f0b4917990e86\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/850e7c9bf2884a48af65231a4ba4c413 : SUCCESS in 10m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/685ea75697e9490184b7f2900d726cfa : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f411ba56bd5141499ba453d53e44c0ff : FAILURE in 1h 00m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c688ea0321d54f03b8216abc841ffcd0 : SUCCESS in 1h 41m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddadf92146274f5bab1e7ade66481bee : SUCCESS in 1h 37m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ace7e4b7f64406d8894bdfeb8cdf30f : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce6708b5d30b4b4c9fbb29fa8687be89 : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ddbff1e84f374d8090079c51b8bdfa6e : SUCCESS in 1h 16m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ad777568a21f4c42ad6d173219503ecf : FAILURE in 34m 06s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e85cadff6f8f48b98b66f1c3fbb64082 : SUCCESS in 1h 54m 17s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81f93b3237994824a4d5158693ab1c36 : SUCCESS in 59m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef374228f62e4ada9dc2d652497b86b2 : SUCCESS in 57m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e2a3b9ea10412d8db486142b4f1209 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2cd51741e339eadb5b9600e1e58689a84c74abca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:05:58.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"510dfd93322cd83b88e42e720fa97ff387514536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-28 00:06:02.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible-os_tempest for 886165,2","accounts_in_message":[],"_revision_number":2},{"id":"fdf8dc2dfe6445486aeb1f0dfd3a625d37316c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:06:28.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2efebe682650c4f1569a9423370c247d8001d70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:10:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"17e668c238d1d7deeb657b7aceb93ba1fbbf3f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-28 00:10:22.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible-os_tempest for 886165,4","accounts_in_message":[],"_revision_number":4},{"id":"75a4460b131c206acee83723bc2afdfe77f8c184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:10:36.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2180f2e3351df9333e34b7518d6183a535199fd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:10:41.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"04d6b8909367b1ff5d71eedec923cb4f27667e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:15:00.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"2d474f6c4abc30f09b88b676c552826ac92c67cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 00:16:14.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"06544dd9ea82f07590d0d815bf78fab05a88da3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-28 02:28:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efd514ecf7554fb7a1edf0b7ca906f6b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fec86e5bfca74e7cbfefe4ce87430879 : SUCCESS in 7m 19s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c8b9e132971f4d63b537f3dd511faec6 : SUCCESS in 59m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4c8fe20f3cf24a728c4ab0c17447f0f9 : SUCCESS in 1h 19m 00s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/988b4a46698147bc99d21198a5230c70 : SUCCESS in 46m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e857c47ae4e44f01a4ff8463bf10ebd2 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bcd16a1cab5b495dbc8d4a0bf521bbba : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a6dd1bf4f7b40c6bc56882864a66985 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17f55f095be946ddbe426b3b312c8815 : SUCCESS in 1h 23m 34s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aa22b9a268714c5aa792d8d4d511e086 : SUCCESS in 1h 29m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8e55094b0d6948fbb861a29c41e7fd13 : SUCCESS in 58m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45424235c52f4a418b690d4c9552dba1 : SUCCESS in 54m 25s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64e934ec722846ffa12d9e6fcf4d9839 : SUCCESS in 2h 05m 20s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb401b8b8cc44b0899e1737025314895 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c6fbf04b590b47c7a03c6575c441d12c : FAILURE in 8m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f16c0ce8933c473fb475623b8e55c6b5 : FAILURE in 55m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dec3ad840d549f88bc09488cf121004 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":8},{"id":"e1ea7c490ce3ae353b913236ebc42d10aa8c6550","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-28 16:53:51.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8a98c8d6a91d0272d2e0d5c280079b7a1fc32422","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-11 23:44:12.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"95b10c30ada9a5d25d63ddb655bbb307c97530b9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-01-12 08:21:09.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"168ec9ce64bc8089ac80368ecde3b1aa048d4bdb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-01-12 09:59:15.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"004f2a1f4256be0d8dcfb882ac07cf42bf66a56d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 10:01:16.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b6109178abeb50d4c737c4651af190db27c607d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:30:08.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14a61848fa4c4a3e8fdc6569c0987465\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fa1d322f36040738703ab933a26ab79 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d0664e8f9454304bf04beec4589ebfe : FAILURE in 16m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3f68c0a9bdc2425c92d6d7c69d00cb90 : FAILURE in 33m 14s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c5df3baf75eb4008a045b4d1105bd598 : SUCCESS in 1h 24m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/46f6bda75acb4a57bf36bd4de5fb8856 : FAILURE in 15m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e1648d5fbf7946fbab4c1f0c1931727c : FAILURE in 14m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc0ed27566b1488d84c9d6b304726f5f : SUCCESS in 44m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ab43b2763e04c45a1d20e9cfe6a242f : SUCCESS in 48m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31f6193927304081b6d6a3c91f9a4bc8 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":8},{"id":"4896bbdb9fc3d5c5c774997d9b14d037adad82b6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 18:17:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2651c868a2376ee6c327c7beb9c469fae28c1e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 21:12:30.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fafa4abcabe3466fb96bfd994516f527\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/593d5482cbde4beb92f8832d86fa2585 : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/dbfd089f735e41c8aadf87440b851620 : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/646fd357472e460ebbc6a3ed96309551 : SUCCESS in 1h 35m 01s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0cbcf10b77a844209fb64468738e6ca7 : SUCCESS in 1h 28m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/181761e4257544019fa460ed26b35f35 : SUCCESS in 43m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f099ac13cf6049ce98721744a6adbf29 : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42bf8f5fa410433e9aade409e0aac891 : SUCCESS in 1h 33m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a285dd918f44516a842e0c91d28c62f : SUCCESS in 1h 43m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/052b03b4e7ea44b38debe4bb3b35e81b : SUCCESS in 1h 02m 04s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/af48e1fc0c98488685b71095481e509d : SUCCESS in 2h 01m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c9d8751575e4250a39d72eca61bf0b9 : SUCCESS in 2h 09m 42s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ac7c039d7fc4fbd8d3b97e49822dde1 : SUCCESS in 2h 04m 29s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3da124931c28431eafb3ba63b52854ee : SUCCESS in 2h 26m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e011ac66ccc4007802d4205e7576805 : FAILURE in 10m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/356a7cdb88c44202a30701aa46a2f025 : FAILURE in 1h 00m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c542a3ebc0427fb096f36189384f67 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":8},{"id":"fb92f8dd1e7d1ba5674efaf41bd556eb280dfd55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 21:14:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"67e8a3e1545ed3b605b23b0d3a91c03dff896ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:27:11.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7132a3491543ad95433c02ad4711a2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64befdbd975046babf0c8b6132a6696a : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/34aa0a32c6dd47e4856ec3c049128da3 : SUCCESS in 47m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/90e2dd545c4e4bc6a292bc70dded1769 : SUCCESS in 49m 44s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3988dc62eb624d9389c8f4e2bab51036 : SUCCESS in 47m 18s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/697cf339950048f58087026098f03814 : SUCCESS in 49m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6dac455e4f7f4f2ab8c6fa556f49d78f : SUCCESS in 1h 12m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ae61afd6b1747c7ac0b27f1f2055c27 : SUCCESS in 43m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c911532b5114c3bac102d5ccf7e9206 : SUCCESS in 49m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e795f5f17684c698dc72c88cb025123 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":8},{"id":"908d9cfdc4768c48a963960181ec6e9a0d9b063a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:27:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"2519efa86750e90e920b7a16de2971e7f6665ecf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:29:02.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327d3e8beffd4504a2be734111a3cded\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9525d7be84447f4acd9d3bc04df15bc : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8585e439083577a7f2160c370602953b8e33c298","revisions":{"e3c52bf07f404df007fb801e121c32a6e8babeea":{"kind":"REWORK","_number":1,"created":"2023-06-15 09:08:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/886165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/1"}}},"commit":{"parents":[{"commit":"6081d1daacf1c4e541d9b97402cb30fc25c625c9","subject":"Merge \"Remove tripleo CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6081d1daacf1c4e541d9b97402cb30fc25c625c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e3c52bf07f404df007fb801e121c32a6e8babeea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e3c52bf07f404df007fb801e121c32a6e8babeea"}]},"branch":"refs/heads/master"},"7242f0e2b0063d1f821ed8823b3311124c113ff6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-28 00:05:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/2"}}},"commit":{"parents":[{"commit":"6081d1daacf1c4e541d9b97402cb30fc25c625c9","subject":"Merge \"Remove tripleo CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6081d1daacf1c4e541d9b97402cb30fc25c625c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:05:58.000000000","tz":0},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/7242f0e2b0063d1f821ed8823b3311124c113ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/7242f0e2b0063d1f821ed8823b3311124c113ff6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e1ba925605027651c7caf6dfab0ecbd095de7d85":{"kind":"REWORK","_number":3,"created":"2025-12-28 00:06:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/3"}}},"commit":{"parents":[{"commit":"6081d1daacf1c4e541d9b97402cb30fc25c625c9","subject":"Merge \"Remove tripleo CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6081d1daacf1c4e541d9b97402cb30fc25c625c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:06:23.000000000","tz":0},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e1ba925605027651c7caf6dfab0ecbd095de7d85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/e1ba925605027651c7caf6dfab0ecbd095de7d85"}]},"branch":"refs/heads/master"},"4feaecc1c508d27ed5fb5b39cac1358ee879e1e1":{"kind":"REWORK","_number":4,"created":"2025-12-28 00:10:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/4"}}},"commit":{"parents":[{"commit":"6081d1daacf1c4e541d9b97402cb30fc25c625c9","subject":"Merge \"Remove tripleo CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6081d1daacf1c4e541d9b97402cb30fc25c625c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:10:15.000000000","tz":0},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/4feaecc1c508d27ed5fb5b39cac1358ee879e1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/4feaecc1c508d27ed5fb5b39cac1358ee879e1e1"}]},"branch":"refs/heads/master"},"bfe0d3ef63354c153937ab446f053ad70b04efe2":{"kind":"REWORK","_number":5,"created":"2025-12-28 00:10:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/5"}}},"commit":{"parents":[{"commit":"6081d1daacf1c4e541d9b97402cb30fc25c625c9","subject":"Merge \"Remove tripleo CI jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/6081d1daacf1c4e541d9b97402cb30fc25c625c9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:10:34.000000000","tz":0},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bfe0d3ef63354c153937ab446f053ad70b04efe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/bfe0d3ef63354c153937ab446f053ad70b04efe2"}]},"branch":"refs/heads/master"},"ff0cad7aeca5877443984a52e7acc6b27b878b01":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-12-28 00:10:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/6"}}},"commit":{"parents":[{"commit":"5d288219bfe1871de4fc96dffc9c8e8e260fde0c","subject":"Merge \"Fix(tempest): Register tempest_router_details to resolve undefined variable _add_router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5d288219bfe1871de4fc96dffc9c8e8e260fde0c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:10:41.000000000","tz":0},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ff0cad7aeca5877443984a52e7acc6b27b878b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/ff0cad7aeca5877443984a52e7acc6b27b878b01"}]},"branch":"refs/heads/master","description":"Rebase"},"311e3a85e67a0658ce4013af38e695dff956a926":{"kind":"REWORK","_number":7,"created":"2025-12-28 00:15:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/7"}}},"commit":{"parents":[{"commit":"5d288219bfe1871de4fc96dffc9c8e8e260fde0c","subject":"Merge \"Fix(tempest): Register tempest_router_details to resolve undefined variable _add_router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5d288219bfe1871de4fc96dffc9c8e8e260fde0c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:15:00.000000000","tz":0},"subject":"Update to cirros 0.6.2","message":"Update to cirros 0.6.2\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/311e3a85e67a0658ce4013af38e695dff956a926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/311e3a85e67a0658ce4013af38e695dff956a926"}]},"branch":"refs/heads/master"},"8585e439083577a7f2160c370602953b8e33c298":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-12-28 00:16:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/65/886165/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/65/886165/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/65/886165/8"}}},"commit":{"parents":[{"commit":"5d288219bfe1871de4fc96dffc9c8e8e260fde0c","subject":"Merge \"Fix(tempest): Register tempest_router_details to resolve undefined variable _add_router\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/5d288219bfe1871de4fc96dffc9c8e8e260fde0c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-06-15 09:08:19.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-28 00:16:14.000000000","tz":0},"subject":"Update CirrOS to latest versions","message":"Update CirrOS to latest versions\n\nhttps://github.com/cirros-dev/cirros/releases\n\n0.5.2 -\u003e 0.5.3\n0.6.2 -\u003e 0.6.3\n\nChange-Id: Ia53e458bf29b0c29480a8036c0387c7ac9997688\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8585e439083577a7f2160c370602953b8e33c298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/8585e439083577a7f2160c370602953b8e33c298"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
