)]}'
{"id":"openstack%2Fopenstack-ansible-os_tempest~907886","triplet_id":"openstack%2Fopenstack-ansible-os_tempest~master~Ibcf16abab20eadf984ad1c8307ad14b43a889793","project":"openstack/openstack-ansible-os_tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-02-06 17:02:50.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-02-06 15:21:05.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ibcf16abab20eadf984ad1c8307ad14b43a889793","subject":"Add missed openstack_resources_image condition","status":"MERGED","created":"2024-02-05 16:55:15.000000000","updated":"2024-02-06 17:07:05.000000000","submitted":"2024-02-06 17:02:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907886","meta_rev_id":"70a79c7386d923059932585889fb8accb40d7efc","_number":907886,"virtual_id_number":907886,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-06 17:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-02-06 12:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-06 15:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-06 15:21:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-05 20:17:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 20:29:50.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-02-06 12:48:07.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"83d0c74481875a9172a1533718a7e36b85d8ef6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-05 16:55:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"888c000820a41935a15bcf8e64d169f7c3b52ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:17:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea0dc7fc70842e892e43599317a5dab\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/035052d4f6924044a47db5126784a53f : SUCCESS in 12m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b817cfee7ae4f289562cfea4898c491 : SUCCESS in 1h 02m 13s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/3e6de50835a44d6c8edf7a8c38967376 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a92a47493ff4c84ae7eb99fabc5186f : SUCCESS in 1h 27m 24s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b726b167ceef44588684036fb3883244 : SUCCESS in 1h 24m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9840440d0971493e99c3972dbc359eb2 : SUCCESS in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67cf9277e636410b88f75cd6c92dd730 : SUCCESS in 1h 37m 53s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7862349ff4347f8b4c8d88a3ff3203b : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41c4ef2c01a54db8af51dd77eeedab44 : SUCCESS in 1h 20m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb73b23af81b4be482f3a66e314d44ce : SUCCESS in 1h 55m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d060473fda9944508c3492ada0381522 : SUCCESS in 1h 39m 07s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be1e145f4f6e442e80d79ed721aa3fa5 : SUCCESS in 2h 07m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a93b01329b5643a79f1e9e26d62000e4 : SUCCESS in 1h 31m 05s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c5e0f1dc3714c2e8fbae2519793b507 : SUCCESS in 1h 03m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/124073075f404ce5907890f72d0360bb : SUCCESS in 57m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ddbcf8933c24053856db963ef9a1001 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cc77dedb1544248f802a197c08f6ca391a1b04e8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-02-05 20:29:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c99a830e51b02cdc759780e903cdf5e6aaf6fdb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-06 08:23:06.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f0ae5fd16a33b115c7609f6df80ffd87114a7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-06 09:35:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c73254b8c262d8b30577c33d59d27e31b30941ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 11:52:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27903e30e08b4de58675986c3456a73d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdb28212ae854cf1b39155171530cf54 : SUCCESS in 12m 33s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6254b6fec7da4b3085a331d815b1d468 : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/68c91d446ee343afb9b8b6c645995a69 : SUCCESS in 1h 22m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/05e2fdfc57c044718a9f8c680782883e : SUCCESS in 1h 28m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e1491a58baf8435a9acac54fed6ac864 : SUCCESS in 1h 32m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c0a6d621e4545d9b241cf05f6ef0d6e : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5be5c8d074334ea997360c0422addfba : SUCCESS in 1h 32m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2e9f98ca7fc47949f29b3d5b46fbfab : SUCCESS in 1h 20m 01s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35faa18fbdae4825a18a067f79da0381 : SUCCESS in 1h 30m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9f1c41e878e34c8d8764e7cebf209d71 : SUCCESS in 1h 49m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5967c9c952504d479de806173c05fcfc : SUCCESS in 1h 57m 40s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/279b92d17871467e9d7c062293aab07b : SUCCESS in 2h 12m 00s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a976f9fb1fbe475db82ead28ee037171 : SUCCESS in 1h 49m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c56c233caddf45939669bf3bb47db9fe : SUCCESS in 1h 12m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f27549d775594ae4869dc9b12714408a : SUCCESS in 51m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f29ed65f1e4da6b9d00d2b75a4ce1a : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2b8785fb6db28bec08309aa862b43404c8d579fd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-02-06 12:48:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c895742adee9163c29594d27ca1b0983f6cea167","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-02-06 15:21:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8d0d4580e98eaf58b40b92d6fd3dfaa916cde76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 15:23:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdf6f555930e1870e02952360556dfc8c3ce9f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:02:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1ad1cc4a8c54d5f805650c5a8f2b162\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0b7b8315d874850a5c580c26cf7dc29 : SUCCESS in 17m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b6076ffe4df44cf1b83b4e388403a82f : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8ad9922afbb1430b906291cc026a5626 : SUCCESS in 1h 20m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32247ece61eb4be993f611845f40fb82 : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/619ab2bb4bac4fd688211745365f81c8 : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cf1a0e6989749e6a6d6b12e6c26d768 : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52ae4e807428404cb78446dfb01e7d38 : SUCCESS in 1h 32m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b6e2571d8e3493ebb5dcf8bb66ed6ac : SUCCESS in 1h 21m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60c9d881843454695d61f3fb4dd3a73 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3f6a291126e6cdaa478fab96b4c038a9414548e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:02:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"70a79c7386d923059932585889fb8accb40d7efc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 17:07:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47013860401042cbbee03807c7a35669\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d0785c3457475c906a4edaa77b4421 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bcb0f20d869999d2fce6f02c4a1cef8882575dd","revisions":{"2e1282fc705ebb9ef4546320b687e23550942579":{"kind":"REWORK","_number":1,"created":"2024-02-05 16:55:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/907886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/86/907886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/1"}}},"commit":{"parents":[{"commit":"56d86e29bdd2e0bbc20d24345dbdc515142cd50e","subject":"Merge \"Adopt for usage openstack_resources role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/56d86e29bdd2e0bbc20d24345dbdc515142cd50e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-05 16:55:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-05 16:55:06.000000000","tz":60},"subject":"Add missed openstack_resources_image condition","message":"Add missed openstack_resources_image condition\n\nIn some jobs where we run tempest we do not have all required services\nGood example of that are infra jobs where only keystone is deployed.\nDuring conversion to usage of openstack_resources role, a condition for\nglance images creation task was missed, that avoided image uploads\nwhen glance is not present.\n\nChange-Id: Ibcf16abab20eadf984ad1c8307ad14b43a889793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2e1282fc705ebb9ef4546320b687e23550942579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/2e1282fc705ebb9ef4546320b687e23550942579"}]},"branch":"refs/heads/master"},"0bcb0f20d869999d2fce6f02c4a1cef8882575dd":{"kind":"REWORK","_number":2,"created":"2024-02-06 09:35:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/907886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_tempest","ref":"refs/changes/86/907886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_tempest refs/changes/86/907886/2"}}},"commit":{"parents":[{"commit":"56d86e29bdd2e0bbc20d24345dbdc515142cd50e","subject":"Merge \"Adopt for usage openstack_resources role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/56d86e29bdd2e0bbc20d24345dbdc515142cd50e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-05 16:55:06.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-06 09:35:44.000000000","tz":60},"subject":"Add missed openstack_resources_image condition","message":"Add missed openstack_resources_image condition\n\nIn some jobs where we run tempest we do not have all required services\nGood example of that are infra jobs where only keystone is deployed.\nDuring conversion to usage of openstack_resources role, a condition for\nglance images creation task was missed, that avoided image uploads\nwhen glance is not present.\n\nChange-Id: Ibcf16abab20eadf984ad1c8307ad14b43a889793\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0bcb0f20d869999d2fce6f02c4a1cef8882575dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_tempest/commit/0bcb0f20d869999d2fce6f02c4a1cef8882575dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
