)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~606787","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ibddd61b7a9e9b78905d548b5ab00e64277449114","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"bug/1801657","hashtags":[],"change_id":"Ibddd61b7a9e9b78905d548b5ab00e64277449114","subject":"fix tox python3 overrides","status":"MERGED","created":"2018-09-30 07:04:33.000000000","updated":"2019-04-24 18:58:49.000000000","submitted":"2019-04-24 18:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"606787-1556132329787-b16e9fe7","meta_rev_id":"12f34ea742911279cf47371b9b54e12633826d81","_number":606787,"virtual_id_number":606787,"owner":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-04-24 18:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"recommended":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"all":[{"value":1,"date":"2019-01-14 08:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-04-24 18:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-01-14 09:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2019-01-12 03:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":1,"date":"2019-04-24 18:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-24 07:46:42.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-01-14 08:19:39.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-01-14 09:13:36.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-04-24 18:50:00.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-04-24 18:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf8e14c15331630d9ff25c11db45b8939a50e54d","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2018-09-30 07:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1211b7aac25a4f5010b22c5da235be25599dc5c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-30 07:47:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/606787/1/check/build-openstack-sphinx-docs/027eddf/html/ : SUCCESS in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/87/606787/1/check/build-openstack-releasenotes/9bd774f/html/ : SUCCESS in 2m 18s\n- openstack-ansible-linters http://logs.openstack.org/87/606787/1/check/openstack-ansible-linters/63afc60/ : SUCCESS in 5m 51s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/606787/1/check/openstack-ansible-functional-centos-7/5035229/ : SUCCESS in 33m 41s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/87/606787/1/check/openstack-ansible-functional-opensuse-150/86fffa8/ : SUCCESS in 38m 32s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/87/606787/1/check/openstack-ansible-functional-opensuse-423/5c1079c/ : SUCCESS in 29m 58s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/87/606787/1/check/openstack-ansible-functional-ubuntu-bionic/d7bab70/ : SUCCESS in 28m 29s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/87/606787/1/check/openstack-ansible-functional-ubuntu-xenial/e8fe645/ : SUCCESS in 27m 25s\n- openstack-tox-docs http://logs.openstack.org/87/606787/1/check/openstack-tox-docs/e4e8f27/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5f54276b5cb7749e4e4ed065efa144481e115206","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-10-07 09:23:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6515177e8b748c6742a79af48fe3593e294a742b","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2018-11-26 07:50:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d2c5104f180cdcc415749a5a0d1fbd1b47bf8872","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2018-11-26 07:51:05.000000000","message":"Topic set to bug/1801657","accounts_in_message":[],"_revision_number":2},{"id":"49d72a60ff6f2ada5a354803f68b5f3d02a59455","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-26 08:31:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/606787/2/check/openstack-ansible-linters/40d7e51/ : SUCCESS in 6m 13s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/606787/2/check/openstack-ansible-functional-centos-7/158ac0f/ : SUCCESS in 34m 23s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/87/606787/2/check/openstack-ansible-functional-opensuse-150/89a9d40/ : SUCCESS in 39m 35s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/87/606787/2/check/openstack-ansible-functional-opensuse-423/0ad8741/ : SUCCESS in 28m 59s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/87/606787/2/check/openstack-ansible-functional-ubuntu-bionic/be7ead8/ : SUCCESS in 27m 13s\n- openstack-tox-docs http://logs.openstack.org/87/606787/2/check/openstack-tox-docs/1e50ff1/html/ : SUCCESS in 2m 50s\n- build-openstack-releasenotes http://logs.openstack.org/87/606787/2/check/build-openstack-releasenotes/ff31aa7/html/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2f1c8148af0691957c7966397622569632495cb6","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2018-12-24 07:46:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b7e1fc9c5b0aba50022ae7e0c30893bc9cf21ccc","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-01-12 03:17:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c99638969230b0c387595d0708abf6f445ebf0e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-12 04:02:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/606787/3/check/openstack-ansible-linters/0f107c1/ : SUCCESS in 4m 40s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/87/606787/3/check/openstack-ansible-functional-centos-7/8a03023/ : SUCCESS in 31m 51s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/87/606787/3/check/openstack-ansible-functional-opensuse-150/c1732cb/ : SUCCESS in 36m 37s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/87/606787/3/check/openstack-ansible-functional-opensuse-423/00d91e1/ : SUCCESS in 32m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/87/606787/3/check/openstack-ansible-functional-ubuntu-bionic/31afbf0/ : SUCCESS in 34m 43s\n- openstack-tox-docs http://logs.openstack.org/87/606787/3/check/openstack-tox-docs/304b80b/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/87/606787/3/check/build-openstack-releasenotes/3bfadc2/html/ : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":3},{"id":"dd71024ad0b0054fcec5b653e7244a6a5913035d","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-01-14 08:19:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9002333269d6b7b504a89d27754628941f61c8ed","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-01-14 09:13:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6c81ac6e2c3e748e8d94b67c79d63c48bc8d01ab","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-24 18:50:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d8df15aa9dfa48a13f46864f87330fef71482db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:50:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"adc147942432e9feca8dc5082ba92f90f5108849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:58:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/606787/3/gate/openstack-ansible-linters/f13692c/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/87/606787/3/gate/openstack-tox-docs/b08d51b/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/87/606787/3/gate/build-openstack-releasenotes/7f64cd0/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"12f34ea742911279cf47371b9b54e12633826d81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:58:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e0314d133906604e64662260098dcf6669b7d8c7","revisions":{"ac52724796f184301138f8372cd1e57ffea13f35":{"kind":"REWORK","_number":1,"created":"2018-09-30 07:04:33.000000000","uploader":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"ref":"refs/changes/87/606787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/87/606787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/1"}}},"commit":{"parents":[{"commit":"626d5cdcd39e9106dde471f199112670fa61a48e","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/626d5cdcd39e9106dde471f199112670fa61a48e"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 07:04:12.000000000","tz":0},"committer":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 07:04:32.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: Ibddd61b7a9e9b78905d548b5ab00e64277449114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ac52724796f184301138f8372cd1e57ffea13f35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ac52724796f184301138f8372cd1e57ffea13f35"}]},"branch":"refs/heads/master"},"6dca48ff779d87c8d0c7c5a2b80a31adfd09c25f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-11-26 07:50:13.000000000","uploader":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"ref":"refs/changes/87/606787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/87/606787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/2"}}},"commit":{"parents":[{"commit":"626d5cdcd39e9106dde471f199112670fa61a48e","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/626d5cdcd39e9106dde471f199112670fa61a48e"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 07:04:12.000000000","tz":0},"committer":{"name":"98k","email":"18552437190@163.com","date":"2018-11-26 07:50:11.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: Ibddd61b7a9e9b78905d548b5ab00e64277449114\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6dca48ff779d87c8d0c7c5a2b80a31adfd09c25f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6dca48ff779d87c8d0c7c5a2b80a31adfd09c25f"}]},"branch":"refs/heads/master"},"e0314d133906604e64662260098dcf6669b7d8c7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-01-12 03:17:33.000000000","uploader":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"ref":"refs/changes/87/606787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/87/606787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/606787/3"}}},"commit":{"parents":[{"commit":"fb2e845256e4aaf4c8aadb943ba3eb77fde6242d","subject":"Merge \"Force force-tlsv12 only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fb2e845256e4aaf4c8aadb943ba3eb77fde6242d"}]}],"author":{"name":"huang.zhiping","email":"huang.zhiping@99cloud.net","date":"2018-09-30 07:04:12.000000000","tz":0},"committer":{"name":"Andriy Shevchenko","email":"huang.zhiping@99cloud.net","date":"2019-01-12 03:17:33.000000000","tz":0},"subject":"fix tox python3 overrides","message":"fix tox python3 overrides\n\nWe want to default to running all tox environments under python 3, so\nset the basepython value in each environment.\n\nWe do not want to specify a minor version number, because we do not\nwant to have to update the file every time we upgrade python.\n\nWe do not want to set the override once in testenv, because that\nbreaks the more specific versions used in default environments like\npy35 and py36.\n\nChange-Id: Ibddd61b7a9e9b78905d548b5ab00e64277449114\nCloses-Bug:  #1801657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e0314d133906604e64662260098dcf6669b7d8c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e0314d133906604e64662260098dcf6669b7d8c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
