)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~654381","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I32f5aa553b172219d570adede4e6bbed28ff69ad","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I32f5aa553b172219d570adede4e6bbed28ff69ad","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 07:16:30.000000000","updated":"2019-04-22 12:11:54.000000000","submitted":"2019-04-22 12:11:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654381-1555935114583-577aa814","meta_rev_id":"4a118ed4b11edf1ff5da8346a0167cf97cb9f1ee","_number":654381,"virtual_id_number":654381,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-04-22 12:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":2,"date":"2019-04-22 10:51:52.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-04-22 10:51:52.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:51:52.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-04-22 12:11:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd07b3b019707b152a8aaff67a47282649782dc5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 07:16:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6e1684ff9b0050274916072441ba5972bcfae15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 09:02:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/81/654381/1/check/openstack-ansible-linters/4714ae5/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/81/654381/1/check/openstack-tox-docs/3fe0906/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/81/654381/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/0bd0ade/ : SUCCESS in 1h 16m 33s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/81/654381/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/79942d7/ : SUCCESS in 1h 01m 23s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/81/654381/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/5b08103/ : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b962ea72de2ab1c9179b9df4f343144fb4152b77","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-04-22 10:51:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9267e1f588e627c4245f5a86ec7b34f98a1a3f93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:52:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"215aa09dd75ac656b043cfcc0d94fb6898651fa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:11:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c53618162d9f4a1c2f005ae95285f7fb172f5a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:11:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/81/654381/1/gate/openstack-ansible-linters/4d430e1/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/81/654381/1/gate/openstack-tox-docs/adbad31/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/81/654381/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/40c6ca4/ : SUCCESS in 1h 13m 08s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/81/654381/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/51af6ef/ : SUCCESS in 57m 37s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/81/654381/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/3730dcf/ : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8c5f92f931fa1c611ec15a387e47d1dad3a6958","revisions":{"e8c5f92f931fa1c611ec15a387e47d1dad3a6958":{"kind":"REWORK","_number":1,"created":"2019-04-22 07:16:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/654381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/81/654381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/654381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/654381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/654381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/654381/1"}}},"commit":{"parents":[{"commit":"9856830f2042ee23838808a1807093ce254fecb4","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9856830f2042ee23838808a1807093ce254fecb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 07:16:28.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I32f5aa553b172219d570adede4e6bbed28ff69ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e8c5f92f931fa1c611ec15a387e47d1dad3a6958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e8c5f92f931fa1c611ec15a387e47d1dad3a6958"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
