)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~654920","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fstein~I73a2576fc02373592f2dd47079c22124771c0a93","project":"openstack/openstack-ansible-os_trove","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I73a2576fc02373592f2dd47079c22124771c0a93","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-23 09:24:53.000000000","updated":"2019-04-24 12:18:20.000000000","submitted":"2019-04-24 12:18:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654920-1556108300067-4f1731f4","meta_rev_id":"b25ff8040f9d6fbeea965f7e3eb04a3e269951ac","_number":654920,"virtual_id_number":654920,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-24 12:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-24 09:15:02.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-23 10:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-23 10:45:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:45:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-24 09:15:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 12:18:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51b562857c71339d6e5a2e7fcd2d9c57056faa42","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-23 09:24:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"162a0d6eaa269700923f8668da19337e46f26236","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 10:45:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5050eb1cb047806498c40d13bd24ec60a36d2f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:58:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/654920/1/check/openstack-ansible-linters/61d4dd2/ : SUCCESS in 5m 33s\n- openstack-tox-docs http://logs.openstack.org/20/654920/1/check/openstack-tox-docs/da2807a/html/ : SUCCESS in 2m 56s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/6289386/ : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/a40ead5/ : SUCCESS in 54m 58s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/6824d7e/ : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f517a9d14f799e4532d0323e42ffe90d562bfa0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:58:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a84e975707fc4a16e7b081b21b8df2579664c7c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:11:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/20/654920/1/gate/openstack-ansible-linters/bdbf95d/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/20/654920/1/gate/openstack-tox-docs/cdc08fd/html/ : SUCCESS in 2m 24s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 finger://ze08.openstack.org/04bb05b0337f49b6b18ee8088ec58660 : RETRY_LIMIT in 32m 39s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable finger://ze03.openstack.org/5a74badc27764fa9aba0d0cdf43a0829 : RETRY_LIMIT in 27m 24s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic finger://ze08.openstack.org/064d6d106b3746bf917fdf2b16e9f367 : RETRY_LIMIT in 50m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a256513eb51ec99f3078687d2fbfe28317a89ac3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:23:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d795941d11dad2b34a66bb4bc3156e97afe0b95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:16:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/20/654920/1/check/openstack-ansible-linters/7371988/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/20/654920/1/check/openstack-tox-docs/336412b/html/ : SUCCESS in 2m 34s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/57d64ca/ : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/dcf98da/ : FAILURE in 10m 06s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/8223af1/ : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"65f898af18158fab814020f9dfdf5b2805ab88da","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 05:55:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"943d7d85bb317c3026d41a1eef53e5306700fd6b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 08:50:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7014aefbf9127b60f9ef763f9c0ca75fe85e1c35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:03:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/20/654920/1/check/openstack-ansible-linters/2f5d072/ : SUCCESS in 4m 55s\n- openstack-tox-docs http://logs.openstack.org/20/654920/1/check/openstack-tox-docs/67cebce/html/ : SUCCESS in 2m 38s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/a578a93/ : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/55e47c3/ : SUCCESS in 1h 17m 28s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/274b182/ : TIMED_OUT in 3h 01m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ada5a41bb009375cc542691c1a72d940f4ed47e1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-24 09:15:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3efff16b0448227f55523bcf23a26f26a3597e4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 10:41:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/654920/1/check/openstack-ansible-linters/6a57aa4/ : SUCCESS in 5m 39s\n- openstack-tox-docs http://logs.openstack.org/20/654920/1/check/openstack-tox-docs/e4b7028/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/2b8fc3f/ : SUCCESS in 1h 21m 24s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/22f8eb0/ : SUCCESS in 55m 51s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/20/654920/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/c9ebd93/ : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3379c87e0868befbb674fc398cb266942c721fd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 10:42:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62e23340b0601ea05fd291e89be557e457ab3ffe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 12:18:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/20/654920/1/gate/openstack-ansible-linters/1f8e04b/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/20/654920/1/gate/openstack-tox-docs/c720274/html/ : SUCCESS in 2m 52s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/20/654920/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/4155bc5/ : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/20/654920/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/f0e186a/ : SUCCESS in 56m 14s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/20/654920/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/1807e9d/ : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7fc7555ec61805b0f53a39b4b9c6d7e631a338ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 12:18:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"267fc327b01f3646cad9011657b2c2e2bf06268a","revisions":{"267fc327b01f3646cad9011657b2c2e2bf06268a":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:24:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/654920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/20/654920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/20/654920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/20/654920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/20/654920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/20/654920/1"}}},"commit":{"parents":[{"commit":"a4583bf9554d49a1ccdf92c43324f36698d4fab3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a4583bf9554d49a1ccdf92c43324f36698d4fab3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:24:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:24:52.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I73a2576fc02373592f2dd47079c22124771c0a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/267fc327b01f3646cad9011657b2c2e2bf06268a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/267fc327b01f3646cad9011657b2c2e2bf06268a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
