)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~656777","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ib12f291002914cc2fcd9df34fad54ef23176ee70","project":"openstack/openstack-ansible-os_trove","branch":"master","hashtags":[],"change_id":"Ib12f291002914cc2fcd9df34fad54ef23176ee70","subject":"Replace Module","status":"MERGED","created":"2019-05-02 13:40:26.000000000","updated":"2019-05-07 23:48:02.000000000","submitted":"2019-05-07 23:48:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656777-1557272882629-bb67493d","meta_rev_id":"ca856511d08932426b10f2e26105c9ccfc6eed8d","_number":656777,"virtual_id_number":656777,"owner":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-07 23:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-05-07 22:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2019-05-07 22:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-03 20:33:55.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-05-06 20:40:27.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-07 22:32:34.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-07 23:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59285929bcbcfcf5b669629f27e804181d46505a","author":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"date":"2019-05-02 13:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19655a9e743962b9a9aeed963a2453789b4b3f69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:06:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/656777/1/check/openstack-ansible-linters/f001e4e/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/77/656777/1/check/openstack-tox-docs/2ea9120/html/ : SUCCESS in 2m 49s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/77/656777/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/b67b85f/ : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/77/656777/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/b535210/ : SUCCESS in 1h 02m 05s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/77/656777/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/da6c025/ : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":1},{"id":"82df61614e02484875dd25e1e66772a5a5e6eb16","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-05-03 20:33:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nalso, not sure if this is going to work. Please, as soon as you have the zuul results, check in ara logs if the task is executed properly","accounts_in_message":[],"_revision_number":1},{"id":"46213087a53a7d7c3d2d29bb0cd697294c06362c","author":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"date":"2019-05-04 10:57:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7dc68d6b50cf13552e8bc493ef3d494c77cf4c36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 12:14:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/656777/2/check/openstack-ansible-linters/026ce17/ : SUCCESS in 7m 47s\n- openstack-tox-docs http://logs.openstack.org/77/656777/2/check/openstack-tox-docs/5bf7386/html/ : SUCCESS in 3m 08s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/77/656777/2/check/openstack-ansible-deploy-aio_metal_trove-centos-7/1a9916e/ : SUCCESS in 1h 12m 59s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/77/656777/2/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/ef9e59c/ : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/77/656777/2/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/cd6f8cc/ : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c190bf01161e90cdb75dcc642e127f07a533e0f2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-06 20:40:27.000000000","message":"Patch Set 2: Code-Review-1\n\nhttps://docs.ansible.com/ansible/latest/modules/os_project_facts_module.html\n\nI believe this is the module you need to use","accounts_in_message":[],"_revision_number":2},{"id":"fa6624922fd1411f07fccc4e315f7ac6453ab42b","author":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"date":"2019-05-06 23:39:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1b184a206426b5db06ca9f93e8fa92ee517c6f19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 01:00:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/656777/3/check/openstack-ansible-linters/e7fa18f/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/77/656777/3/check/openstack-tox-docs/80bbbf4/html/ : SUCCESS in 2m 58s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/77/656777/3/check/openstack-ansible-deploy-aio_metal_trove-centos-7/a865273/ : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/77/656777/3/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/f225df5/ : SUCCESS in 1h 05m 36s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/77/656777/3/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/9292a83/ : SUCCESS in 1h 09m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8e5571fa75125eb4fa9c510cc47763d911a3740b","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-07 22:32:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7c3e06320a28eb97ba22b714ce4b8a9d8d31c501","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-07 22:32:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2c9d58af1d741ce1f4a4552e190e855d78aa1c12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:32:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48e03f11f2a514ae02bb9855140efd2f0c2cd6e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 23:48:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/77/656777/3/gate/openstack-ansible-linters/6173cdf/ : SUCCESS in 6m 00s\n- openstack-tox-docs http://logs.openstack.org/77/656777/3/gate/openstack-tox-docs/d5b48d4/html/ : SUCCESS in 2m 58s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/77/656777/3/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/418b634/ : SUCCESS in 1h 14m 42s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/77/656777/3/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/830f14d/ : SUCCESS in 1h 08m 20s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/77/656777/3/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/abbaea6/ : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ca856511d08932426b10f2e26105c9ccfc6eed8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 23:48:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2563c39ab0820cb71d5624d6d35147bb77aebd39","revisions":{"6447fc215d0dc0bf2bdb29a8ced13ee71a9d950b":{"kind":"REWORK","_number":1,"created":"2019-05-02 13:40:26.000000000","uploader":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"ref":"refs/changes/77/656777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/77/656777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/1"}}},"commit":{"parents":[{"commit":"bc9c52c8ab645b14ebed21e0b07b75b51a1b499f","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/bc9c52c8ab645b14ebed21e0b07b75b51a1b499f"}]}],"author":{"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","date":"2019-05-02 13:37:41.000000000","tz":330},"committer":{"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","date":"2019-05-02 13:37:41.000000000","tz":330},"subject":"Replace Module","message":"Replace Module\n\nModified the file to use the os_project module instead\nof the keystone module.\n\nw\n\nChange-Id: Ib12f291002914cc2fcd9df34fad54ef23176ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6447fc215d0dc0bf2bdb29a8ced13ee71a9d950b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6447fc215d0dc0bf2bdb29a8ced13ee71a9d950b"}]},"branch":"refs/heads/master"},"d5cfc3122d6dd0cb9ae401d7f4b6d41b7bcf4bab":{"kind":"REWORK","_number":2,"created":"2019-05-04 10:57:46.000000000","uploader":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"ref":"refs/changes/77/656777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/77/656777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/2"}}},"commit":{"parents":[{"commit":"bc9c52c8ab645b14ebed21e0b07b75b51a1b499f","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/bc9c52c8ab645b14ebed21e0b07b75b51a1b499f"}]}],"author":{"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","date":"2019-05-02 13:37:41.000000000","tz":330},"committer":{"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","date":"2019-05-04 10:57:11.000000000","tz":330},"subject":"Replace Module","message":"Replace Module\n\nModified the file to use the os_project module instead\nof the keystone module.\n\nw\n\nChange-Id: Ib12f291002914cc2fcd9df34fad54ef23176ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d5cfc3122d6dd0cb9ae401d7f4b6d41b7bcf4bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d5cfc3122d6dd0cb9ae401d7f4b6d41b7bcf4bab"}]},"branch":"refs/heads/master"},"2563c39ab0820cb71d5624d6d35147bb77aebd39":{"kind":"REWORK","_number":3,"created":"2019-05-06 23:39:36.000000000","uploader":{"_account_id":30324,"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","username":"kanishk.godha"},"ref":"refs/changes/77/656777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/77/656777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/77/656777/3"}}},"commit":{"parents":[{"commit":"bc9c52c8ab645b14ebed21e0b07b75b51a1b499f","subject":"Merge \"fix tox python3 overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/bc9c52c8ab645b14ebed21e0b07b75b51a1b499f"}]}],"author":{"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","date":"2019-05-02 13:37:41.000000000","tz":330},"committer":{"name":"Kanishk Godha","email":"kanishkg@vexxhost.com","date":"2019-05-06 23:39:16.000000000","tz":330},"subject":"Replace Module","message":"Replace Module\n\nModified the file to use the os_project module instead\nof the keystone module.\n\nw\n\nChange-Id: Ib12f291002914cc2fcd9df34fad54ef23176ee70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
