)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~658042","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ie0a83f428d75c4c5f8c6e14df9084c21da53275e","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie0a83f428d75c4c5f8c6e14df9084c21da53275e","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:35:42.000000000","updated":"2019-06-27 13:42:35.000000000","submitted":"2019-06-27 13:42:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658042-1561642955773-98f816f4","meta_rev_id":"f4d7b93b76ff5334925b8cf6582156966ecbeabd","_number":658042,"virtual_id_number":658042,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-27 13:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-06-27 12:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-06-27 12:05:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 07:13:47.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-05-20 20:22:33.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-20 20:36:17.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-06-18 18:53:40.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-18 19:50:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-27 12:05:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-27 13:42:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bcb4998d9b53cd709ebe7af677dd15056e725171","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:35:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15666d1b07fd19656ec8fc9ccc50943c57898382","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:56:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6291a872c671239a4ffe7870f4f2588f39c0cf9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:01:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/1/check/openstack-ansible-linters/cc7bdb8/ : SUCCESS in 8m 18s\n- openstack-tox-docs http://logs.openstack.org/42/658042/1/check/openstack-tox-docs/218f63a/html/ : SUCCESS in 2m 35s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/1/check/build-openstack-releasenotes/817af25/html/ : SUCCESS in 2m 31s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/e6873ef/ : FAILURE in 1h 07m 35s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/3c6addd/ : FAILURE in 58m 10s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/5aaf860/ : FAILURE in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e4db87ba543ad22bab054219c9df86cf7e0bda66","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-05-10 07:13:47.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5155a23a7f278c23d774dc65ea2eaf613cdc0c04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 08:31:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/1/check/openstack-ansible-linters/5ce81b2/ : SUCCESS in 6m 14s\n- openstack-tox-docs http://logs.openstack.org/42/658042/1/check/openstack-tox-docs/ae00d7d/html/ : SUCCESS in 3m 14s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/1/check/build-openstack-releasenotes/8df22a4/html/ : SUCCESS in 2m 20s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/8fd0115/ : FAILURE in 1h 12m 41s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/632882a/ : FAILURE in 59m 10s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/9a5868f/ : FAILURE in 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5b9165cc6b51fcef7dfd34bd892c9ad8105c2f81","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-10 16:51:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce46768bc3bdc45eb7d9cf9572788d03b50740d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 18:53:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/1/check/openstack-ansible-linters/4e65db3/ : SUCCESS in 8m 06s\n- openstack-tox-docs http://logs.openstack.org/42/658042/1/check/openstack-tox-docs/b2bbfbd/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/1/check/build-openstack-releasenotes/3097034/html/ : SUCCESS in 2m 22s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/484a2e8/ : FAILURE in 1h 53m 12s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/9620492/ : FAILURE in 48m 17s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/ec9b562/ : FAILURE in 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ce684019dd91250750df710aa2b85898ee95701c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-20 20:22:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1bc21c4998dc30e0bf4860f02ed75775e1b6c2e4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-20 20:36:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"72d9c5fa731b753994b67870a3c7382fedba2d9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 21:32:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/1/check/openstack-ansible-linters/73a435a/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/42/658042/1/check/openstack-tox-docs/528fc69/html/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/1/check/build-openstack-releasenotes/4a9362b/html/ : SUCCESS in 2m 48s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/b06a179/ : FAILURE in 1h 01m 57s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/572f923/ : FAILURE in 51m 30s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/e03bee5/ : FAILURE in 54m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ae1021476cdb140da6582be9809698490a01be0a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 04:42:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44a563cc8d999494b7fe3dc13229bfd9ee419f3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 05:56:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/1/check/openstack-ansible-linters/9fe7cda/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/42/658042/1/check/openstack-tox-docs/99e993a/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/1/check/build-openstack-releasenotes/5f09280/html/ : SUCCESS in 3m 14s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/daa4e99/ : FAILURE in 1h 10m 37s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/182aa18/ : FAILURE in 54m 53s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/bac9876/ : FAILURE in 46m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d84aab00b95175ce67866a8aa310dc686a796eb1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:19:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ded3f70cd0db23fc061721756f28b8c9077caf46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 23:35:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/2/check/openstack-ansible-linters/7c5a769/ : SUCCESS in 8m 59s\n- openstack-tox-docs http://logs.openstack.org/42/658042/2/check/openstack-tox-docs/89ebe52/html/ : SUCCESS in 3m 45s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/2/check/build-openstack-releasenotes/753385f/html/ : SUCCESS in 4m 22s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/2/check/openstack-ansible-deploy-aio_metal_trove-centos-7/da270ea/ : FAILURE in 1h 07m 52s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/2/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/0803b07/ : FAILURE in 58m 04s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/2/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/5d998fa/ : FAILURE in 49m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6aabd4ca91fb274c8b7508e7ca9b2d0b09c20f02","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:16:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2567604437d418c8e9e88d98d932d69421176a86","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-18 18:53:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2ac42c6c516b10e8961737b95aa22dcfdead0fb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:49:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/3/check/openstack-ansible-linters/faafbe2/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/42/658042/3/check/openstack-tox-docs/9a14b58/html/ : SUCCESS in 3m 20s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/3/check/build-openstack-releasenotes/f79554f/html/ : SUCCESS in 2m 37s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/3/check/openstack-ansible-deploy-aio_metal_trove-centos-7/d649b5e/ : FAILURE in 1h 05m 50s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/3/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/911f3e4/ : FAILURE in 42m 17s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/3/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/1cee921/ : FAILURE in 55m 21s","accounts_in_message":[],"_revision_number":3},{"id":"687730ec21c44e0a371fd2407ad0f47a15081639","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-18 19:50:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"476d424c4da8e0691784a001626c03d5713fc107","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:10:11.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/3/check/openstack-ansible-linters/d732fed/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/42/658042/3/check/openstack-tox-docs/16cd0e1/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/3/check/build-openstack-releasenotes/bdefcd7/html/ : SUCCESS in 3m 01s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/3/check/openstack-ansible-deploy-aio_metal_trove-centos-7/8ce8e36/ : FAILURE in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/3/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/2067f1b/ : FAILURE in 49m 57s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/3/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/a794d8e/ : FAILURE in 48m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bc19dc7955a80b7fcdd41215f15a4075f098b9a1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-19 11:13:55.000000000","message":"Removed Workflow+1 by Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"5135db29ea1aa049709459676d1991bf6ee0d141","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-27 10:23:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"08474c6b30964a0dde5557d7363f572f920bed63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:54:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/4/check/openstack-ansible-linters/6b2c8b7/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/42/658042/4/check/openstack-tox-docs/d173274/html/ : SUCCESS in 2m 21s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/4/check/build-openstack-releasenotes/2e36b89/html/ : SUCCESS in 2m 35s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/4/check/openstack-ansible-deploy-aio_metal_trove-centos-7/571125f/ : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/4/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/881c73e/ : SUCCESS in 1h 09m 32s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/4/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/5e0fb66/ : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b8c31a12e8d26db121b3768d59d8cbdbaf916e70","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-27 12:05:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ff41c67149174e4ad877704958689b273d659bb7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:06:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"47c0c2d9be18ea1338de33358ad7df560a392638","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:42:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/42/658042/4/gate/openstack-ansible-linters/00b7137/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/42/658042/4/gate/openstack-tox-docs/540130e/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/42/658042/4/gate/build-openstack-releasenotes/cd23773/html/ : SUCCESS in 3m 10s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/42/658042/4/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/3616e43/ : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/42/658042/4/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/02b5969/ : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/42/658042/4/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/7e823ed/ : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":4},{"id":"5dcc8f228ef857aaddfeb2d8349f9d3125d334d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:42:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"10399145d387438af238ae87ccc0f013132a3b83","revisions":{"dfed9065bdeacc20b034ccd0697762f7ec023f3e":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:35:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/658042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/42/658042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/1"}}},"commit":{"parents":[{"commit":"2563c39ab0820cb71d5624d6d35147bb77aebd39","subject":"Replace Module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:35:41.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie0a83f428d75c4c5f8c6e14df9084c21da53275e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/dfed9065bdeacc20b034ccd0697762f7ec023f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/dfed9065bdeacc20b034ccd0697762f7ec023f3e"}]},"branch":"refs/heads/master"},"eb120fc5e83a74004f6031c83fb3fdf7916aa98c":{"kind":"REWORK","_number":2,"created":"2019-06-11 22:19:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/658042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/42/658042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/2"}}},"commit":{"parents":[{"commit":"2563c39ab0820cb71d5624d6d35147bb77aebd39","subject":"Replace Module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:19:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:19:39.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie0a83f428d75c4c5f8c6e14df9084c21da53275e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/eb120fc5e83a74004f6031c83fb3fdf7916aa98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/eb120fc5e83a74004f6031c83fb3fdf7916aa98c"}]},"branch":"refs/heads/master"},"3b7150f37297608a68c97cc80642827954313524":{"kind":"REWORK","_number":3,"created":"2019-06-18 18:16:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/658042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/42/658042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/3"}}},"commit":{"parents":[{"commit":"2563c39ab0820cb71d5624d6d35147bb77aebd39","subject":"Replace Module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:16:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:16:48.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie0a83f428d75c4c5f8c6e14df9084c21da53275e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3b7150f37297608a68c97cc80642827954313524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3b7150f37297608a68c97cc80642827954313524"}]},"branch":"refs/heads/master"},"10399145d387438af238ae87ccc0f013132a3b83":{"kind":"REWORK","_number":4,"created":"2019-06-27 10:23:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/658042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/42/658042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/658042/4"}}},"commit":{"parents":[{"commit":"951924ab51f78f984cbfa5625ce1f3b708e9fd16","subject":"Use keystone catalog and fix looking up and storing admin tenant id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/951924ab51f78f984cbfa5625ce1f3b708e9fd16"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 10:23:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 10:23:33.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie0a83f428d75c4c5f8c6e14df9084c21da53275e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/10399145d387438af238ae87ccc0f013132a3b83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/10399145d387438af238ae87ccc0f013132a3b83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
