)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~658617","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fstein~I6456179337753a402e71527f158e951066d10967","project":"openstack/openstack-ansible-os_trove","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6456179337753a402e71527f158e951066d10967","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:59:51.000000000","updated":"2019-05-12 03:25:00.000000000","submitted":"2019-05-12 03:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658617-1557631500956-f41f2753","meta_rev_id":"a21d10c807616a0424e17bf5019e0d8bf95cc0ed","_number":658617,"virtual_id_number":658617,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:54:53.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dbc0e01f8ea5de923daf5ab99047b16db64ae3f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:59:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dddd1c6892d271f2fa096ad225786ef25e911081","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:24:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/17/658617/1/check/openstack-ansible-linters/6dc3175/ : SUCCESS in 6m 52s\n- openstack-tox-docs http://logs.openstack.org/17/658617/1/check/openstack-tox-docs/a6c901e/html/ : SUCCESS in 3m 16s\n- build-openstack-releasenotes http://logs.openstack.org/17/658617/1/check/build-openstack-releasenotes/fca650d/html/ : SUCCESS in 4m 00s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/17/658617/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/6c0f5b7/ : SUCCESS in 1h 18m 33s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/17/658617/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/0965227/ : SUCCESS in 58m 56s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/17/658617/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/547f38f/ : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a8c6209969a5c09003da04886a382c1809a59fdc","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:54:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73fb72533ac89ee705eaca50d24281d19176cc3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:55:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"515817002add5885faaa127260e766f2ce7d7741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:25:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"caf602de7c80dd70c09f0e77b03a102cabebb524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:25:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/17/658617/1/gate/openstack-ansible-linters/db9e06a/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/17/658617/1/gate/openstack-tox-docs/8b354f4/html/ : SUCCESS in 2m 12s\n- build-openstack-releasenotes http://logs.openstack.org/17/658617/1/gate/build-openstack-releasenotes/3457865/html/ : SUCCESS in 2m 18s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/17/658617/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/60f636d/ : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/17/658617/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/d20d049/ : SUCCESS in 1h 04m 21s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/17/658617/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/b9e2a52/ : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50c6ed6602303deb4902cb7759ca8c3c8d7acd7a","revisions":{"50c6ed6602303deb4902cb7759ca8c3c8d7acd7a":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:59:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/658617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/17/658617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/17/658617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/17/658617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/17/658617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/17/658617/1"}}},"commit":{"parents":[{"commit":"267fc327b01f3646cad9011657b2c2e2bf06268a","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/267fc327b01f3646cad9011657b2c2e2bf06268a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:59:50.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6456179337753a402e71527f158e951066d10967\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/50c6ed6602303deb4902cb7759ca8c3c8d7acd7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/50c6ed6602303deb4902cb7759ca8c3c8d7acd7a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
