)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~660298","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~If5db29b5def5065acf3480ea9cad3ed0b03c53b8","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"If5db29b5def5065acf3480ea9cad3ed0b03c53b8","subject":"Add integrated coverage","status":"MERGED","created":"2019-05-21 08:57:05.000000000","updated":"2019-07-19 17:00:37.000000000","submitted":"2019-07-19 17:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660298-1563555637081-8d65fae4","meta_rev_id":"d45e148cc923de31acaf80bf28940ae4f3bd2fb8","_number":660298,"virtual_id_number":660298,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-19 17:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 12:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 13:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-07-19 13:30:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 12:57:02.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 13:30:06.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 17:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ef03b10cfca2ddbcae5e39592d4fe78eb2d5baf","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 08:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49c6d0b0b06e1b3c3a02ec421031cb820dc88269","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:11:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/98/660298/1/check/openstack-ansible-linters/6008cf0/ : SUCCESS in 5m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/660298/1/check/openstack-ansible-deploy-aio_metal-debian-stable/fbe03ef/ : FAILURE in 50m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/660298/1/check/openstack-ansible-deploy-aio_metal-centos-7/213d85e/ : FAILURE in 1h 09m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/98/660298/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/3da55a9/ : FAILURE in 1h 04m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/660298/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/710aa92/ : FAILURE in 45m 08s\n- openstack-tox-docs http://logs.openstack.org/98/660298/1/check/openstack-tox-docs/57b1363/html/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3991ef8e2d8d0929d8718f4aa82c0a82ac9c0576","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-13 07:36:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ef80e8cb2b7a1d967949386ad2bb63da0c1c0137","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 09:00:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/98/660298/2/check/openstack-ansible-linters/d26f9d3/ : SUCCESS in 4m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/660298/2/check/openstack-ansible-deploy-aio_metal-debian-stable/4fd9df7/ : SUCCESS in 1h 03m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/660298/2/check/openstack-ansible-deploy-aio_metal-centos-7/12b7960/ : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/98/660298/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/84e2e93/ : SUCCESS in 1h 22m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/660298/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/90c960e/ : SUCCESS in 1h 04m 03s\n- openstack-tox-docs http://logs.openstack.org/98/660298/2/check/openstack-tox-docs/7bf5cff/html/ : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":2},{"id":"45ecb650238405d1e169c842741deb46cd17e89c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 12:57:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a2a391035fa5b97a4ed6209ce6b4a8b092901279","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 13:30:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf58f55fc6fd7a76b28374cacb7f7e3eef4ed9aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:30:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90f0d3d06e76cd75b1c690f06c7969849e2077de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 15:10:01.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"957eede51c006b101ae11b528cfb3595a11bcb08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:00:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/98/660298/2/gate/openstack-ansible-linters/e4d5355/ : SUCCESS in 8m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/98/660298/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/b1b77e0/ : SUCCESS in 1h 11m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/98/660298/2/gate/openstack-ansible-deploy-aio_metal-centos-7/4b883dc/ : SUCCESS in 1h 45m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/98/660298/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/4e12238/ : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/98/660298/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/02efdf0/ : SUCCESS in 1h 06m 11s\n- openstack-tox-docs http://logs.openstack.org/98/660298/2/gate/openstack-tox-docs/38c70c5/html/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a3378fb05cbac23036c5e73633a5f2bc3aa7371c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:00:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"568c0494cfb60f53494ff526070dac46c0eb1f03","revisions":{"3292ee9e45837172e2c603320949c342f79e072b":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:57:05.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/98/660298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/98/660298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/1"}}},"commit":{"parents":[{"commit":"2563c39ab0820cb71d5624d6d35147bb77aebd39","subject":"Replace Module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:56:15.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:56:33.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nChange-Id: If5db29b5def5065acf3480ea9cad3ed0b03c53b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3292ee9e45837172e2c603320949c342f79e072b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3292ee9e45837172e2c603320949c342f79e072b"}]},"branch":"refs/heads/master"},"568c0494cfb60f53494ff526070dac46c0eb1f03":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-13 07:36:24.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/98/660298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/98/660298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/98/660298/2"}}},"commit":{"parents":[{"commit":"e9b5b968c115e0470473450024828a2ff5e84f86","subject":"Merge \"Cap sphinx for py2 to match global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e9b5b968c115e0470473450024828a2ff5e84f86"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:56:15.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-13 07:36:24.000000000","tz":0},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nWith the dynamic execution of the roles, this role\nalso get tested by the integrated repo.\n\nChange-Id: If5db29b5def5065acf3480ea9cad3ed0b03c53b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/568c0494cfb60f53494ff526070dac46c0eb1f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/568c0494cfb60f53494ff526070dac46c0eb1f03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
