)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~660299","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I271c9d26c1a7c37e6b2a2c0126336e8e3bf5dc66","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"I271c9d26c1a7c37e6b2a2c0126336e8e3bf5dc66","subject":"Clarify supported SUSE versions","status":"ABANDONED","created":"2019-05-21 08:58:10.000000000","updated":"2020-01-07 12:07:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bfac478eecce3843254710031d99ccfee685b07b","_number":660299,"virtual_id_number":660299,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-05-21 10:08:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 10:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f421350c2d076a4fec851b385bfe5d419f38ace","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 08:58:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fd3e5635c7b57c73afcc16b94ebd9d7f377b6b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:08:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/99/660299/1/check/openstack-ansible-linters/1123f78/ : SUCCESS in 8m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/660299/1/check/openstack-ansible-deploy-aio_metal-debian-stable/99a5cdb/ : FAILURE in 53m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/660299/1/check/openstack-ansible-deploy-aio_metal-centos-7/c32f7bc/ : FAILURE in 1h 03m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/99/660299/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/bca722e/ : FAILURE in 1h 00m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/660299/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3c7f1bb/ : FAILURE in 16m 03s\n- openstack-tox-docs http://logs.openstack.org/99/660299/1/check/openstack-tox-docs/5e13054/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4da4ec265ecee6f08c3c8114dd5eaf97d311f9b4","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-07 12:07:00.000000000","message":"Abandoned\n\nI am abandoning this, I don\u0027t have enough time to work on this anymore :/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a38d6e1be9440652d54191e909f06f31c970cfd","revisions":{"6a38d6e1be9440652d54191e909f06f31c970cfd":{"kind":"REWORK","_number":1,"created":"2019-05-21 08:58:10.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/99/660299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/99/660299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/99/660299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/99/660299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/99/660299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/99/660299/1"}}},"commit":{"parents":[{"commit":"3292ee9e45837172e2c603320949c342f79e072b","subject":"Add integrated coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3292ee9e45837172e2c603320949c342f79e072b"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:57:33.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 08:57:33.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I271c9d26c1a7c37e6b2a2c0126336e8e3bf5dc66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6a38d6e1be9440652d54191e909f06f31c970cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6a38d6e1be9440652d54191e909f06f31c970cfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
