)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~664348","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Frocky~I6e698848dd5bdbedb69475ede58ec77d616606bd","project":"openstack/openstack-ansible-os_trove","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I6e698848dd5bdbedb69475ede58ec77d616606bd","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:45:22.000000000","updated":"2019-06-10 17:59:25.000000000","submitted":"2019-06-10 17:59:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664348-1560189565586-2d20edbd","meta_rev_id":"e90b79f02159395337f5d7d8f372de691a17dadc","_number":664348,"virtual_id_number":664348,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 17:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:07:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:07:07.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 17:59:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12ab7d6f116151033a5615a2a9173652572d3cb8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35de3782aaa64caef41c2bf4c08405d55d2b20b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:02:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/48/664348/1/check/openstack-ansible-linters/fdbb935/ : SUCCESS in 5m 59s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/48/664348/1/check/openstack-ansible-functional-centos-7/8ba685a/ : SUCCESS in 34m 52s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/48/664348/1/check/openstack-ansible-functional-opensuse-423/a034936/ : FAILURE in 8m 04s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/48/664348/1/check/openstack-ansible-functional-opensuse-150/0641a99/ : FAILURE in 9m 50s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/48/664348/1/check/openstack-ansible-functional-ubuntu-bionic/1c87b6b/ : SUCCESS in 25m 42s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/48/664348/1/check/openstack-ansible-functional-ubuntu-xenial/e68e937/ : SUCCESS in 27m 14s\n- build-openstack-sphinx-docs http://logs.openstack.org/48/664348/1/check/build-openstack-sphinx-docs/0ffdfdc/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"da1382e6eeb4777f43f7c0ddf07632ac4cfed40c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:07:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21953133c75c9174f13042d6c16b4a2e9e78b249","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:07:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e2ed2e86ccfc2040ef853ef59d3d4f28f81e5aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:59:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e153315a0857b2932a0d68ea206ca3b9a924df0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:59:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/48/664348/1/gate/openstack-ansible-linters/c6f03c5/ : SUCCESS in 5m 01s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/48/664348/1/gate/openstack-ansible-functional-centos-7/dbe89c6/ : SUCCESS in 33m 10s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/48/664348/1/gate/openstack-ansible-functional-opensuse-423/611c297/ : FAILURE in 9m 03s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/48/664348/1/gate/openstack-ansible-functional-ubuntu-bionic/46ae283/ : SUCCESS in 30m 47s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/48/664348/1/gate/openstack-ansible-functional-ubuntu-xenial/8ad7708/ : SUCCESS in 34m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/48/664348/1/gate/build-openstack-sphinx-docs/b469de2/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e11a9d8a7e589a4d613e4a100b810f1b13784fcd","revisions":{"e11a9d8a7e589a4d613e4a100b810f1b13784fcd":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:45:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/664348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/48/664348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/664348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/664348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/664348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/664348/1"}}},"commit":{"parents":[{"commit":"d68b960200422db6b4e4291c4f52872ae1183490","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d68b960200422db6b4e4291c4f52872ae1183490"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:45:20.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I6e698848dd5bdbedb69475ede58ec77d616606bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e11a9d8a7e589a4d613e4a100b810f1b13784fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e11a9d8a7e589a4d613e4a100b810f1b13784fcd"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
