)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~667287","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I779ba715d20d83b1efe4f07226a5eadd7e0a1870","project":"openstack/openstack-ansible-os_trove","branch":"master","hashtags":[],"change_id":"I779ba715d20d83b1efe4f07226a5eadd7e0a1870","subject":"Use keystone catalog and fix looking up and storing admin tenant id","status":"MERGED","created":"2019-06-25 08:31:41.000000000","updated":"2019-06-27 03:46:34.000000000","submitted":"2019-06-27 03:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667287-1561607194883-e8d46088","meta_rev_id":"1a9c949ea056237b359bdf6edca6c19627ff2c56","_number":667287,"virtual_id_number":667287,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-06-27 03:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-06-25 20:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-27 01:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-06-27 01:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 08:32:27.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-25 17:17:20.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-25 20:35:34.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-06-27 01:58:35.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-06-27 03:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b494b042fa28f439c02557ee83b5f810bc557fc4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-25 08:31:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4636e83760a280bb30ccd6dcf60c67325c88c3f7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-25 08:32:27.000000000","message":"Patch Set 1: Code-Review+2\n\nNot sure about the state of trove, but I like to see this added, so it\u0027s consistent everywhere.","accounts_in_message":[],"_revision_number":1},{"id":"78fed5bfc686ca2d1bd8aed2a6ecbc7f9e5b142f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 12:14:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/87/667287/1/check/openstack-ansible-linters/733cf44/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/87/667287/1/check/openstack-tox-docs/bb3bc87/html/ : SUCCESS in 2m 51s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 finger://ze06.openstack.org/42e0ab66ae6d411aa036e2763dd1d26e : RETRY_LIMIT in 53m 19s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/87/667287/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/8301ad2/ : RETRY_LIMIT in 12m 02s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic finger://ze10.openstack.org/ccc2fc00216c40e1828e262c7afe5f41 : RETRY_LIMIT in 51m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b3a38882b4ca992a6456e5de1e46e808721b42a0","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-25 12:53:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d71d134929b2dd927c4378472a9fbaecea06bc94","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-25 17:17:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e33a15242c1a3711b2a6c660bd13f39e92dadb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 18:10:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/87/667287/1/check/openstack-ansible-linters/9995f82/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/87/667287/1/check/openstack-tox-docs/ad505b8/html/ : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/87/667287/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/05409ca/ : FAILURE in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/87/667287/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/e92521b/ : FAILURE in 49m 17s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/87/667287/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/5c3a7b9/ : FAILURE in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d5cd7badd220ccd16dcb94115af4bcff12a11847","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-25 18:17:33.000000000","message":"Patch Set 1: Workflow-1\n\nchanging this to WIP as there are apparently some things to fix. Debugging locally","accounts_in_message":[],"_revision_number":1},{"id":"10f7b61d814ea9bd441286e1f6b1bd4c70410dea","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-25 19:00:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1540501e6e09cc139be805c2d801aa5af014e04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:34:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/667287/2/check/openstack-ansible-linters/87e3691/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/87/667287/2/check/openstack-tox-docs/c7562b3/html/ : SUCCESS in 3m 53s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/87/667287/2/check/openstack-ansible-deploy-aio_metal_trove-centos-7/ecf5f6e/ : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/87/667287/2/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/91f6378/ : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/87/667287/2/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/0e152d7/ : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5dd64367113e297e13c527d68209cb3fc12dc1a2","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-06-25 20:35:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"63eddd2e8fdbd1b4e83427bd26fc18bf6a7b54c1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-06-27 01:58:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b7d3a31db278df6ca4fe03fb1bba4a7b8666ecdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 01:58:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d58a75cb0020bb88bb7d36e4b136ef28e909f75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:46:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/87/667287/2/gate/openstack-ansible-linters/7b4884f/ : SUCCESS in 7m 28s\n- openstack-tox-docs http://logs.openstack.org/87/667287/2/gate/openstack-tox-docs/2f2a010/html/ : SUCCESS in 3m 52s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/87/667287/2/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/f068a8b/ : SUCCESS in 1h 42m 08s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/87/667287/2/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/f92afee/ : SUCCESS in 1h 02m 25s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/87/667287/2/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/3478362/ : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1a9c949ea056237b359bdf6edca6c19627ff2c56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"951924ab51f78f984cbfa5625ce1f3b708e9fd16","revisions":{"a3dd0a5e0648f838a667ae2b73756c3e1c46a902":{"kind":"REWORK","_number":1,"created":"2019-06-25 08:31:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/667287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/87/667287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/1"}}},"commit":{"parents":[{"commit":"2563c39ab0820cb71d5624d6d35147bb77aebd39","subject":"Replace Module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-25 08:24:06.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-25 08:30:30.000000000","tz":60},"subject":"Use keystone catalog to retrieve service endpoints","message":"Use keystone catalog to retrieve service endpoints\n\nTrove is now able to properly use the service catalog [1].\n\n[1] https://review.opendev.org/#/c/574254/\n\nChange-Id: I779ba715d20d83b1efe4f07226a5eadd7e0a1870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a3dd0a5e0648f838a667ae2b73756c3e1c46a902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a3dd0a5e0648f838a667ae2b73756c3e1c46a902"}]},"branch":"refs/heads/master"},"951924ab51f78f984cbfa5625ce1f3b708e9fd16":{"kind":"REWORK","_number":2,"created":"2019-06-25 19:00:53.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/87/667287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/87/667287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/667287/2"}}},"commit":{"parents":[{"commit":"2563c39ab0820cb71d5624d6d35147bb77aebd39","subject":"Replace Module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2563c39ab0820cb71d5624d6d35147bb77aebd39"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-25 08:24:06.000000000","tz":60},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-06-25 19:00:52.000000000","tz":0},"subject":"Use keystone catalog and fix looking up and storing admin tenant id","message":"Use keystone catalog and fix looking up and storing admin tenant id\n\nTrove is now able to properly use the service catalog [1].\n\nWe have another[2] patch that this patch depends-on, which\ncauses us to have a circular dependency. As the change is minimal,\nshould be ok to squash them together to be able to fix that role.\n\nThe paragraph bellow contains the commit message of the patch that is\nbeing squashed.\n\nThe \u0027Get admin tenant id\u0027 task was passing invalid arguments to the\nos_project_facts module, and the \u0027Store admin tenant id\u0027 task was not\nusing the correct variable to retrieve the project id.\n\n[1] https://review.opendev.org/#/c/574254/\n[2] https://review.opendev.org/#/c/665458/\n\nChange-Id: I779ba715d20d83b1efe4f07226a5eadd7e0a1870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/951924ab51f78f984cbfa5625ce1f3b708e9fd16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/951924ab51f78f984cbfa5625ce1f3b708e9fd16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
