)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~671331","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ib16362a8becb6666f36e5abdb93c548de2a269da","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"Ib16362a8becb6666f36e5abdb93c548de2a269da","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-17 16:52:43.000000000","updated":"2019-07-18 20:32:53.000000000","submitted":"2019-07-18 20:32:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671331-1563481973758-5870c8eb","meta_rev_id":"3d3e6df1b0e182ed32205efb175d3976eeeeab9d","_number":671331,"virtual_id_number":671331,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-18 16:59:23.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-18 20:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-18 12:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-18 12:52:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-18 12:52:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 12:38:35.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-18 12:52:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-18 20:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c009f8d2a10a4456dbd3c21926b3403e45d6e98","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-17 16:52:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"057972d65ddb0f7ab1c8441b0024c1044495272d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:52:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/671331/1/check/openstack-ansible-linters/31b7249/ : SUCCESS in 7m 03s\n- openstack-tox-docs http://logs.openstack.org/31/671331/1/check/openstack-tox-docs/f05b4dd/html/ : SUCCESS in 3m 06s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/31/671331/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/52bdb52/ : SUCCESS in 1h 48m 04s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/31/671331/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/68dd96f/ : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/31/671331/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/a632762/ : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a3294a01e4c5590e71104abf8f73c3064e7a3548","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 12:38:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dd9d29625056a53e5b562d723743a67ff1034db","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-18 12:52:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"095c0c2fa39753044a6ac89b09d10ccb9295313a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 12:52:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec11a1e920d3100766db926e0407d552de03d67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:57:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/31/671331/1/gate/openstack-ansible-linters/f26a13e/ : SUCCESS in 6m 08s\n- openstack-tox-docs http://logs.openstack.org/31/671331/1/gate/openstack-tox-docs/4aaf1a8/html/ : SUCCESS in 2m 43s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/31/671331/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/24f27a0/ : SUCCESS in 1h 54m 29s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/31/671331/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/50f4a7c/ : SUCCESS in 1h 05m 07s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/31/671331/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/ae5b3ce/ : FAILURE in 40m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b8ac33fb48005d2732d53456847b0cc563029c2a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-18 16:59:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41b064b62d2d817e83bc1b7ebe9269feb494abc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 18:39:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/671331/1/check/openstack-ansible-linters/d620498/ : SUCCESS in 5m 51s\n- openstack-tox-docs http://logs.openstack.org/31/671331/1/check/openstack-tox-docs/01f99b9/html/ : SUCCESS in 3m 05s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/31/671331/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/5536bfa/ : SUCCESS in 1h 37m 59s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/31/671331/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/c9914a6/ : SUCCESS in 1h 11m 33s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/31/671331/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/abd416b/ : SUCCESS in 1h 25m 46s","accounts_in_message":[],"_revision_number":1},{"id":"890a62cdbdfecc38e8115ca0dec6fc5872fe2b70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 18:39:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24ae3ae4a8a4013f3bfacd18e3d005b6881e276c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:32:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/31/671331/1/gate/openstack-ansible-linters/56940db/ : SUCCESS in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/31/671331/1/gate/openstack-tox-docs/61b90c2/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/31/671331/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/e0a14bc/ : SUCCESS in 1h 47m 24s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/31/671331/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/53c3152/ : SUCCESS in 1h 21m 38s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/31/671331/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/1c0e144/ : SUCCESS in 1h 06m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1b31d5759ea7166ccdff39648281966b42727c5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 20:32:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed8d692fa9fe31d01f845f69e35cb4970296161a","revisions":{"ed8d692fa9fe31d01f845f69e35cb4970296161a":{"kind":"REWORK","_number":1,"created":"2019-07-17 16:52:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/31/671331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/31/671331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/31/671331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/31/671331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/31/671331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/31/671331/1"}}},"commit":{"parents":[{"commit":"546efd4b4dc29c3fd720e7d6954007045bc4196c","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/546efd4b4dc29c3fd720e7d6954007045bc4196c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 16:52:24.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 16:52:24.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nChange-Id: Ib16362a8becb6666f36e5abdb93c548de2a269da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ed8d692fa9fe31d01f845f69e35cb4970296161a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ed8d692fa9fe31d01f845f69e35cb4970296161a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
