)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~671334","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I754049a0e37daadddf1fa52a3b651284ee02b27e","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"static_imports","hashtags":[],"change_id":"I754049a0e37daadddf1fa52a3b651284ee02b27e","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-07-17 17:03:37.000000000","updated":"2019-07-18 21:18:08.000000000","submitted":"2019-07-18 21:18:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671334-1563484688603-b2961d89","meta_rev_id":"a92188b1743f1342c8ebef1d421b3995db0c5032","_number":671334,"virtual_id_number":671334,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-18 16:59:37.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-18 21:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-18 12:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-18 12:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-18 12:52:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 12:38:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-18 12:52:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-18 21:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43a46cffa74d748b939840b7512b223b7b475812","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-17 17:03:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54a7e29ca7d7a01f1432a50b5978c5b032a818ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/671334/1/check/openstack-ansible-linters/2f8bbb6/ : SUCCESS in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/34/671334/1/check/openstack-tox-docs/74c5939/html/ : SUCCESS in 2m 50s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/34/671334/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/0924c95/ : SUCCESS in 1h 29m 11s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/34/671334/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/c1ef867/ : SUCCESS in 1h 06m 53s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/34/671334/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/ac3a13f/ : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8cc11aac357cbdabff5ab55f32dc2ff967c43cc6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-18 12:38:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c923093b10ee33b7605196edfe72526c478450df","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-18 12:52:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ebbc0369302cfd7f275ac983fb1cb3700a3beb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 12:52:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74e2f6da450e09fb97cd4d48ed5205655a650d46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 15:03:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/34/671334/1/gate/openstack-ansible-linters/38e4128/ : SUCCESS in 6m 09s\n- openstack-tox-docs http://logs.openstack.org/34/671334/1/gate/openstack-tox-docs/939a8d3/html/ : SUCCESS in 4m 09s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/34/671334/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/72cca7a/ : SUCCESS in 1h 21m 09s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/34/671334/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/3c843ca/ : SUCCESS in 1h 02m 26s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/34/671334/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/f59ce0b/ : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1b3ff5381859ff7a754a418bb1a49e6186ac735d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-18 16:59:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f8fe9d5941e6c6524d1e4404b9ba95791c6a882f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 19:32:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/671334/1/check/openstack-ansible-linters/c823808/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/34/671334/1/check/openstack-tox-docs/ed5654c/html/ : SUCCESS in 2m 45s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/34/671334/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/8390b43/ : SUCCESS in 2h 24m 47s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/34/671334/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/e3b7e4a/ : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/34/671334/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/f8e2cca/ : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b068f8de387631a61a55497b1ba02c83531f335d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 19:32:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47146f27734a965ff5dd31fd23813a16593b2d13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 21:18:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/34/671334/1/gate/openstack-ansible-linters/2b9c2da/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/34/671334/1/gate/openstack-tox-docs/cfa495f/html/ : SUCCESS in 2m 59s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/34/671334/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/afe7f2e/ : SUCCESS in 1h 40m 22s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/34/671334/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/07a8713/ : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/34/671334/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/a7b48df/ : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7cdf2f390933112061ddeacb3955fd3a32ea22e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 21:18:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72f165602ccf32c6df3b8f3b3124ae320e03d984","revisions":{"72f165602ccf32c6df3b8f3b3124ae320e03d984":{"kind":"REWORK","_number":1,"created":"2019-07-17 17:03:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/34/671334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/34/671334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/34/671334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/34/671334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/34/671334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/34/671334/1"}}},"commit":{"parents":[{"commit":"546efd4b4dc29c3fd720e7d6954007045bc4196c","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/546efd4b4dc29c3fd720e7d6954007045bc4196c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 17:03:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 17:03:22.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like trove-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I754049a0e37daadddf1fa52a3b651284ee02b27e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/72f165602ccf32c6df3b8f3b3124ae320e03d984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/72f165602ccf32c6df3b8f3b3124ae320e03d984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
