)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~671741","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fstein~I754049a0e37daadddf1fa52a3b651284ee02b27e","project":"openstack/openstack-ansible-os_trove","branch":"stable/stein","topic":"static_imports-stable/stein","hashtags":[],"change_id":"I754049a0e37daadddf1fa52a3b651284ee02b27e","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-07-19 14:08:26.000000000","updated":"2019-07-19 18:36:52.000000000","submitted":"2019-07-19 18:36:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671741-1563561412293-a91269ad","meta_rev_id":"87623c4bee82349cfeab2157f185919af216132c","_number":671741,"virtual_id_number":671741,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 18:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 16:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 17:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 17:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 16:09:23.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 17:03:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:36:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bf91676f776733f18f38501d4c2ac88230cd0a7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 14:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5eaace6fc169cc8e9b962bd39a07559d36bd7a92","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 16:09:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32c790695880ab87dbe9da7a84613e03a94e80ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:53:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/671741/1/check/openstack-ansible-linters/cc39072/ : SUCCESS in 7m 24s\n- openstack-tox-docs http://logs.openstack.org/41/671741/1/check/openstack-tox-docs/878c95e/html/ : SUCCESS in 5m 22s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/41/671741/1/check/openstack-ansible-deploy-aio_metal_trove-centos-7/e5c8d47/ : SUCCESS in 1h 36m 21s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/41/671741/1/check/openstack-ansible-deploy-aio_metal_trove-debian-stable/88984a4/ : SUCCESS in 1h 02m 14s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/41/671741/1/check/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/0068913/ : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d34e697529afeb75c375a690fc2eb482d6ed5343","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 17:03:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"096f92c33cf44d5eb01b319b23bb6d1ca3f1c05e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:03:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d879ba94a8180ea687b314f22a3e1c1177ea119f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:36:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3272e4482a7dd202ee6d9b88650e254caa0cd4c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:36:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/41/671741/1/gate/openstack-ansible-linters/a215a7c/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/41/671741/1/gate/openstack-tox-docs/0b25183/html/ : SUCCESS in 4m 27s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 http://logs.openstack.org/41/671741/1/gate/openstack-ansible-deploy-aio_metal_trove-centos-7/13c9611/ : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable http://logs.openstack.org/41/671741/1/gate/openstack-ansible-deploy-aio_metal_trove-debian-stable/868fded/ : SUCCESS in 1h 04m 08s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic http://logs.openstack.org/41/671741/1/gate/openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic/6fb6377/ : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d456e24f7469c3072bfe81bbd06630afc8c517e5","revisions":{"d456e24f7469c3072bfe81bbd06630afc8c517e5":{"kind":"REWORK","_number":1,"created":"2019-07-19 14:08:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/671741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/41/671741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/41/671741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/41/671741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/41/671741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/41/671741/1"}}},"commit":{"parents":[{"commit":"08fc2deea6a06b850d8c002c7dd8bf208a74e9ef","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/08fc2deea6a06b850d8c002c7dd8bf208a74e9ef"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-17 17:03:22.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-19 14:08:14.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like trove-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: I754049a0e37daadddf1fa52a3b651284ee02b27e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d456e24f7469c3072bfe81bbd06630afc8c517e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d456e24f7469c3072bfe81bbd06630afc8c517e5"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
