)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~671830","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I2cbbbda11ed2d141ef3bc86e552040c766fe55bd","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2cbbbda11ed2d141ef3bc86e552040c766fe55bd","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-19 18:27:14.000000000","updated":"2019-07-20 16:36:03.000000000","submitted":"2019-07-20 16:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671830-1563640563651-a91269ad","meta_rev_id":"bc68bcc8c979d5f6b4fed46c7c09b24974cdf0a7","_number":671830,"virtual_id_number":671830,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-20 16:36:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-20 16:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-20 16:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-20 16:25:38.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-20 16:36:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f0b77bed7f51b1989ee0922807216a0e918b349","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-19 18:27:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92640f2b758b72aa4312c913b82e4b9494e7d70c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:49:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/30/671830/1/check/requirements-check/caa417f/ : SUCCESS in 2m 57s\n- openstack-ansible-linters http://logs.openstack.org/30/671830/1/check/openstack-ansible-linters/39d6e41/ : SUCCESS in 6m 52s\n- openstack-tox-docs http://logs.openstack.org/30/671830/1/check/openstack-tox-docs/b8b8d92/ : POST_FAILURE in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/30/671830/1/check/build-openstack-releasenotes/d332cdb/html/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1504dc8d86d3a28936707488523d76cd2153accd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-20 09:38:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0636cf082e2f9709add347002aa8f05eb07c1ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 09:46:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/30/671830/2/check/requirements-check/fe51557/ : SUCCESS in 2m 31s\n- openstack-ansible-linters http://logs.openstack.org/30/671830/2/check/openstack-ansible-linters/7172afc/ : SUCCESS in 5m 38s\n- openstack-tox-docs http://logs.openstack.org/30/671830/2/check/openstack-tox-docs/107123d/html/ : SUCCESS in 2m 24s\n- build-openstack-releasenotes http://logs.openstack.org/30/671830/2/check/build-openstack-releasenotes/bcbcc58/html/ : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"dbbbb1e5ad087218c04ed41e537f3e516289d9cb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-20 16:25:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e9f3631f3a99c9386af904a8a2345b4b84ce7950","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:25:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f817069cc58d2505f891b1c25ed519c660fef747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:36:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a0ca43e3662cad78199f83995e3186319de048c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:36:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/30/671830/2/gate/requirements-check/4346f80/ : SUCCESS in 1m 48s\n- openstack-ansible-linters http://logs.openstack.org/30/671830/2/gate/openstack-ansible-linters/37a1bbe/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/30/671830/2/gate/openstack-tox-docs/3725c99/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/30/671830/2/gate/build-openstack-releasenotes/42e29cc/html/ : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd2dbfa87c96232964954aad6d7b5d390dfd4004","revisions":{"3c0e8a9f2522689802ddddbad115d22d4f0de9b4":{"kind":"REWORK","_number":1,"created":"2019-07-19 18:27:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/671830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/30/671830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/1"}}},"commit":{"parents":[{"commit":"131d3a8cdaa54c1b301742b05827bf67c02a87b3","subject":"Merge \"Add integrated coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/131d3a8cdaa54c1b301742b05827bf67c02a87b3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:27:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:27:10.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2cbbbda11ed2d141ef3bc86e552040c766fe55bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3c0e8a9f2522689802ddddbad115d22d4f0de9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3c0e8a9f2522689802ddddbad115d22d4f0de9b4"}]},"branch":"refs/heads/master"},"fd2dbfa87c96232964954aad6d7b5d390dfd4004":{"kind":"REWORK","_number":2,"created":"2019-07-20 09:38:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/671830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/30/671830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/30/671830/2"}}},"commit":{"parents":[{"commit":"131d3a8cdaa54c1b301742b05827bf67c02a87b3","subject":"Merge \"Add integrated coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/131d3a8cdaa54c1b301742b05827bf67c02a87b3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:38:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:38:52.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2cbbbda11ed2d141ef3bc86e552040c766fe55bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fd2dbfa87c96232964954aad6d7b5d390dfd4004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fd2dbfa87c96232964954aad6d7b5d390dfd4004"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
