)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~694360","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I6c23d9de8f1ee9cae016204af9ec4be025de580d","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"I6c23d9de8f1ee9cae016204af9ec4be025de580d","subject":"Replace git.openstack.org with opendev.org","status":"MERGED","created":"2019-11-14 17:06:51.000000000","updated":"2019-11-21 05:58:23.000000000","submitted":"2019-11-21 05:57:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694360-1574315838657-ca8dfbed","meta_rev_id":"d5f8faa42d99a3477ed41499b51fcfdb68191432","_number":694360,"virtual_id_number":694360,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-20 18:37:30.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 05:57:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-11-18 16:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 10:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-11-18 16:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 10:05:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-18 16:48:24.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-11-21 05:57:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f377f492119dbf90beb6305294be92d01c4d559d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-14 17:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a4d65c0ca9554ccca3d2209d5107c80839b62a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 19:44:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6de91b444f064a27adf0c106484a13fe : SUCCESS in 5m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fdcabbf46cab4c99a8a5e65fafb09d19 : SUCCESS in 1h 35m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8d8961ce571f4641923dd65d09538a61 : SUCCESS in 1h 45m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bf9f374b309649f5a3e6f31cf091a8bd : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e0ba624bb094f1a92dc9500b5805e95 : SUCCESS in 1h 05m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e691a74cb36749f6967220c4b7f59ce2 : SUCCESS in 2h 32m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e503dcc745f48328ae86a1f865c375e : SUCCESS in 1h 52m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1caf7469ccd46f19fd86712a2095f56 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7072b8716e774f2ebf727d3707513d9b : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"91eb53a40fe575a7d4c06dc8db4e526e9ebefd7a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 10:05:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd5ae273f4f012df1d0d8b5b9ad83aa814d9f958","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-11-18 16:48:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7fad3e72a065a1920e8feb9929c6f4f5a7374377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:48:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cb283323202deddd72fd061541be71bd3cf41ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 20:06:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/978c4831cc0d44d6a20c511b53af6f66 : SUCCESS in 5m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2b4425d361f34495b2be76399d31c630 : SUCCESS in 1h 21m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/24b65840a1904b2382bffe104783cdbb : SUCCESS in 1h 53m 10s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1b4fbe20dd354dbbb134462e43a89edb : SUCCESS in 2h 10m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c591fd8a917483296555f47d0bd61ba : SUCCESS in 1h 16m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/54a770a7ac4e46aa9e54f4f25967cedd : TIMED_OUT in 3h 02m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daad4f18e1e848c993943af3c099db7e : SUCCESS in 1h 57m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e64a0e66a547b2a22e4b255aab16ad : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa24b068eb4b48e98d0292691bf15f6d : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"65f23d30f12ae2ebabbe23f983806d72972b5227","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 21:57:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c93debb1548dccb9c8d142a675e1a093c92fc5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 01:04:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/09de9552c40e497ab8c9741d4dd12bd7 : SUCCESS in 5m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d8313a1750b644cdb874b47ddb7cb474 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2ec764731568405b8878ffab9d5e4098 : SUCCESS in 1h 45m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bcfa95cc9a8f4ac5bc0c0d1036f6ae32 : SUCCESS in 1h 42m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18ead6dab83d4d47bbd6565007a6c92e : SUCCESS in 1h 15m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57fbd2f7c28146c5bf65b701375cac0a : TIMED_OUT in 3h 02m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/621aca1a48d342c3bc2d340f3f143e69 : SUCCESS in 2h 03m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9dbf23f8cc42fcb96cefdec1a04183 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f18c7ee59f04da1b6c9d83240696dd3 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c91e2ad76ef0cb571a8cddd145b5dd87f35f837b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 17:51:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc858ab6baf0fec423f65c578d2d01ca1cf1bf9a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-20 18:37:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"30dc26fe6464898cc03b126d087a6ada03c2847b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 20:44:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/934f82dd8fff4030b6a62c54075d6e91 : SUCCESS in 9m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4d17e3543cbd4d1d80e8ef3aa722a1e6 : SUCCESS in 1h 24m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/84664c5d4e8f48a4b502843ed1e995ba : SUCCESS in 1h 42m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/15e6c9352004440e8c5a974ce688504f : SUCCESS in 1h 29m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/542cbf01dbeb4de088facdd1b146dff4 : SUCCESS in 1h 11m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ce2ff19eb75f464d9479cbddaeec3942 : SUCCESS in 2h 40m 12s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21f1e4db776e4cf88ff6ba518a4947a9 : FAILURE in 1h 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db94b113b3574ebaa3462c5ca72bb75b : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/367338dd9e774deebf8bf339235689ad : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d07ab0ded597989a376ba8f0bd1b86fdc01fab60","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 22:18:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1000958568e27ff57e2171abbecfb8219282358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 00:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7f54c5b15cfc401d9d1a42e3d567fd40 : SUCCESS in 5m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/634837d7866f49d6ab0c49d60037a2ae : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/909524abaaa5438ebffe8129fdbcb9b4 : SUCCESS in 1h 47m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/81f646f2d5e84108a4696c0f2b68f1e7 : SUCCESS in 1h 29m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4677ceb6649c40b285fd723cba9adfdb : SUCCESS in 1h 17m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b54c04b7eab5435ab6d95088690a5735 : SUCCESS in 2h 10m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4f2b7c605044e799a42bdaf151d28b4 : SUCCESS in 1h 40m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4547ae5601141ab95506e9790bd7ba4 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82c99080bb584942945c1c3a881fa394 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"22002d024f466674a477499c2a1f8a2389e48b0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 00:37:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43cf5bf7d2e527b593c5bd5e564e78c30dd3872f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 05:57:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2ae1628c603b82131b8327ad1e3765e070dd443b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 05:57:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/778e9a01f6d949ff96094d91a9ef18f8 : SUCCESS in 5m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4dffd70c93574a96be716647950fc466 : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e9a0d67eff04414daa81ed17f2fc2e53 : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5a3f75d82f34494fb4da0d7e2feb3bc9 : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20e7ba2e1bbb4923bde7c3aa61240e25 : SUCCESS in 1h 10m 42s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d5c8fac544504ac39ce6bd35af82f056 : SUCCESS in 2h 41m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/781446ecb1fb4162ab11f5cead870e36 : SUCCESS in 1h 43m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752aa889648e49b098b8d5da54bff9dc : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83c2695ba7d2456bb5d76d2f4f46fc39 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7817e88806110b1196555e63dbf8e8654a04dec7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 05:58:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdfe19bb14aa43f08ccff88289a3587d : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f276a81da25b4471a77468af80388f3d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2e13b08eed894602615c2b28fb0a19690a8c53d","revisions":{"b2e13b08eed894602615c2b28fb0a19690a8c53d":{"kind":"REWORK","_number":1,"created":"2019-11-14 17:06:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/60/694360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/60/694360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/60/694360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/60/694360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/60/694360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/60/694360/1"}}},"commit":{"parents":[{"commit":"9c6059c514838fb3f15429fc00378dbe7aac5e9f","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9c6059c514838fb3f15429fc00378dbe7aac5e9f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 17:06:25.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-11-14 17:06:30.000000000","tz":120},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\nAlso we change upper constraints url due to [1]\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nDepends-On: https://review.opendev.org/693841/\nChange-Id: I6c23d9de8f1ee9cae016204af9ec4be025de580d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/b2e13b08eed894602615c2b28fb0a19690a8c53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/b2e13b08eed894602615c2b28fb0a19690a8c53d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
