)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~695447","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Ftrain~Ib25705c08787164d2f070cc4acbdc0837bc30c8b","project":"openstack/openstack-ansible-os_trove","branch":"stable/train","topic":"bug/1738529-stable/train","hashtags":[],"change_id":"Ib25705c08787164d2f070cc4acbdc0837bc30c8b","subject":"Fix keystone API call","status":"MERGED","created":"2019-11-21 14:32:07.000000000","updated":"2019-11-21 22:43:59.000000000","submitted":"2019-11-21 22:42:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695447-1574376173568-752aeac9","meta_rev_id":"2422412558516b85e52c415eafa5d0b768749649","_number":695447,"virtual_id_number":695447,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 22:42:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-11-21 15:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-21 15:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":2,"date":"2019-11-21 20:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-11-21 20:15:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 15:29:09.000000000","updated_by":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"reviewer":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"state":"REVIEWER"},{"updated":"2019-11-21 15:42:52.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-11-21 22:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fa54cc9ce0287c675e372919a3fda377ab3e2f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 14:32:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3bc243ca2987aecf0d9b1d6d21340d0279592783","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 14:32:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d9fed6ebe0c376d0420fa666403b4731af4a6de2","accounts_in_message":[],"_revision_number":1},{"id":"d4ded991ddda52f0adbd9172d6de928f22bb608c","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-11-21 15:29:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a08051060b75defb4156faef87a6348ca65608d","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-11-21 15:42:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e12db7248ebf57ef844bde1d480bad8e8d4bb08a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:31:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/22cde00d73f0431da5ef2228d94e415f : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ec81ec3eabbd44eeaee9cd218bbde834 : SUCCESS in 1h 17m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c95232943d8d4644a9c29801ba2b787c : SUCCESS in 1h 48m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0442de2a06124e41921df4d8294c54db : SUCCESS in 1h 26m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0357d189c1f147d9a38f72c913b78716 : SUCCESS in 1h 10m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40cbcc7d7c13401881460f76501f36e0 : SUCCESS in 2h 52m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d22cc1d7034b431e8e458a0db4cf8a14 : SUCCESS in 1h 58m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be9ca498f06544a19ae93cd6c9fefc14 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6f6910359a9462944428245146745f9576614726","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 20:15:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc6a81ea036be97caa2a0a91a768f70f64954012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 20:15:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88e7dfe9de0b165f20424711421916454083f26b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 22:42:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d4cd6e1b24df15e41a106b3dab7636d793ba8f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 22:42:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f236e39fd68c49208c12463aa5e473b1 : SUCCESS in 4m 51s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3a9af92bf2074c2b8c0a3493a0c3f126 : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f9dd5b7412c94a639e078f45f20ccd73 : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f3cd2c430c954cbfbe9ad218d3b31a3d : SUCCESS in 1h 17m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2a6c21896894f4da0f7b07242275f9d : SUCCESS in 1h 13m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e48c0be6d182458aa114bdd7878c9bbf : SUCCESS in 2h 23m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67826ae03ca4415c8db16288099ff3df : SUCCESS in 1h 46m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d5f52ae19b484b922e5867e02cfb42 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b312e9a15d68ec3d41c0ae1ff5937a7c8ff8d819","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 22:43:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be4847ba8ee14fe1a59be6718c5f9707 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8330de8cb6163d9ef2250f3855c87cebc3d621ed","revisions":{"8330de8cb6163d9ef2250f3855c87cebc3d621ed":{"kind":"REWORK","_number":1,"created":"2019-11-21 14:32:07.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/47/695447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/47/695447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/695447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/695447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/695447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/695447/1"}}},"commit":{"parents":[{"commit":"25823d8e8268392c93320d2512bd3e4a6f8d3151","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/25823d8e8268392c93320d2512bd3e4a6f8d3151"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-11-13 14:00:39.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-11-21 14:32:07.000000000","tz":0},"subject":"Fix keystone API call","message":"Fix keystone API call\n\nAPI call should use trove_service_net_endpoint_type\n\nChange-Id: Ib25705c08787164d2f070cc4acbdc0837bc30c8b\nClose-bug: #1738529\n(cherry picked from commit 04aa9a527ab50b6370db2a26459c355d55d67d3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/8330de8cb6163d9ef2250f3855c87cebc3d621ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/8330de8cb6163d9ef2250f3855c87cebc3d621ed"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
