)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~695448","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fstein~Ib25705c08787164d2f070cc4acbdc0837bc30c8b","project":"openstack/openstack-ansible-os_trove","branch":"stable/stein","topic":"bug/1738529-stable/train-stable/stein","hashtags":[],"change_id":"Ib25705c08787164d2f070cc4acbdc0837bc30c8b","subject":"Fix keystone API call","status":"MERGED","created":"2019-11-21 14:32:15.000000000","updated":"2019-11-21 17:32:36.000000000","submitted":"2019-11-21 17:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695448-1574357465429-ca8dfbed","meta_rev_id":"7ba1b800dfb75f2f8a96c238254cc633aa37fafc","_number":695448,"virtual_id_number":695448,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 17:31:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2019-11-21 15:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-21 15:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":2,"date":"2019-11-21 16:11:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-11-21 16:11:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 15:29:02.000000000","updated_by":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"reviewer":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"state":"REVIEWER"},{"updated":"2019-11-21 15:43:00.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2019-11-21 17:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1e5a943f94e4f40b374379638bce870d5d12659","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 14:32:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"a414730de8c8d883ca534672e39ded9061194a9b","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-11-21 15:29:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b383b559ad5ebfdf064b9f9312c6e8b766b14633","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-11-21 15:43:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ef61927027ea1448692dc073806adbc0807c8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:08:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dc81ecabd7574e7bbc52848c8d2b97fc : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce38f82f301481b9790cca3fad83026 : SUCCESS in 3m 04s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 https://zuul.opendev.org/t/openstack/build/cbffe210ac804826b9510047402631e9 : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable https://zuul.opendev.org/t/openstack/build/48c77ae56c9242d8a5cb19d90f03ef28 : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfe47b1273174545a927ef02f8bd4f79 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"17606ae70dec3df3c9a60b7462361eee6081a83c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 16:11:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b3205b4e4c1393c1d33aab2a60b20a28faff5d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 16:11:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"738ece4ecc16c95d3c8c5bd0b4040508ac13ba2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb9882414fb657a269494905220f78523167f502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:31:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c9fc15967783439a9c664ab553a991e0 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd9bba51c0140bd9fb901dab956a69c : SUCCESS in 3m 34s\n- openstack-ansible-deploy-aio_metal_trove-centos-7 https://zuul.opendev.org/t/openstack/build/f4d545de00f848c882db29b2b50586d6 : SUCCESS in 1h 18m 36s\n- openstack-ansible-deploy-aio_metal_trove-debian-stable https://zuul.opendev.org/t/openstack/build/c10b2c015f7c42059823196a3a12ea70 : SUCCESS in 1h 04m 00s\n- openstack-ansible-deploy-aio_metal_trove-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/618b455ef1d044cfa8544ec895873387 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b2c4b6c2bd4ecaad0d4107dbb2233a2b55839bc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 17:32:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ba1cc148bf40e28b64b055f3fd79dd : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9fed6ebe0c376d0420fa666403b4731af4a6de2","revisions":{"d9fed6ebe0c376d0420fa666403b4731af4a6de2":{"kind":"REWORK","_number":1,"created":"2019-11-21 14:32:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/48/695448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/48/695448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/695448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/695448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/695448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/48/695448/1"}}},"commit":{"parents":[{"commit":"63718a4b3adac3668c7c3bc8f16d0ba988a9283c","subject":"Switch to opensuse-15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/63718a4b3adac3668c7c3bc8f16d0ba988a9283c"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-11-13 14:00:39.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-11-21 14:32:15.000000000","tz":0},"subject":"Fix keystone API call","message":"Fix keystone API call\n\nAPI call should use trove_service_net_endpoint_type\n\nChange-Id: Ib25705c08787164d2f070cc4acbdc0837bc30c8b\nClose-bug: #1738529\n(cherry picked from commit 04aa9a527ab50b6370db2a26459c355d55d67d3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d9fed6ebe0c376d0420fa666403b4731af4a6de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d9fed6ebe0c376d0420fa666403b4731af4a6de2"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
