)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~727332","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I88fc0d606ca0bffd4b2e44e3b2202955c38d4b36","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I88fc0d606ca0bffd4b2e44e3b2202955c38d4b36","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-05-12 17:52:52.000000000","updated":"2020-05-13 16:04:54.000000000","submitted":"2020-05-13 16:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727332-1589385728650-80453ba6","meta_rev_id":"5f3895e3a91b5bb8418dac145fb0536fdcb048f3","_number":727332,"virtual_id_number":727332,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 16:02:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-12 21:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-12 21:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 21:19:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-13 16:02:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0930b3ceb0e4a11559151821c30291d9d0cb688","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 17:52:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66c70acc3a3d90d88d86523499cce98f0f258870","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 18:05:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07cf60f2ec36b401e6e50eb5ad9593b8cabfcc32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 21:16:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ab79e18285de40a9bc5432c9c18be2bb : SUCCESS in 5m 23s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d0d75c1a37da425a907c51595964926a : SUCCESS in 11m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b28a84917c1a45e8bd6364bd94d44ec3 : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a69afebcce034ad28cedb8b063ee5f75 : SUCCESS in 1h 51m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bd2d8e8aae6c4421a48c0aff2c4834f7 : SUCCESS in 1h 24m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74d877c788e541afabc79d3b9c7ee0fd : SUCCESS in 1h 16m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/64de47a29d3041929a32052791f15e2e : SUCCESS in 2h 36m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6601268493334485ab21542cbcd97ca7 : SUCCESS in 2h 00m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89f33389bcc4bc6aebb9035d6a70e3b : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3921d489a1534c3f8520370ffc33af12 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3eab179ff0754d0862c30328943440d64e839bd1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 21:19:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0ec5daee788f9a1e9a072b0a810e92070711c0af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 21:19:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28ebfc95aba658fe00a3e955f2ee5f76dad52f91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:09:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a02627773204dbb9cbbc936339aa543 : SUCCESS in 4m 13s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97d423ecc4b5469ebb83c1e215158c64 : SUCCESS in 9m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/083ae039e84d40fd94aa24ea24e3d83c : FAILURE in 1h 17m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fa608fa59ead4f8aaf11043ff2b43f1a : SUCCESS in 1h 38m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/be2e8447c1b847d7ae680f57f4211658 : SUCCESS in 1h 29m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db781cae27614f0183ad5d07bc039548 : SUCCESS in 1h 22m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1a1b118dad48179b6c46c4d7fb5ca7 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/946047cd8aff46369d2cb6b4030cd026 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"fc6d6c9c56141880caf49dcfabc026f94129255a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-13 07:22:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd19b8b4edc10f5f0904e22ebe3d753a1501d0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 10:00:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/08297672b030444d93ac4716c17b39ac : SUCCESS in 2m 47s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/312205bc9a304a59ae44cfa8259e5443 : SUCCESS in 9m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a788b12c2cfc4b71ac7b1ce44a18e562 : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aa5b9ae6553349fc84045ea8b5e5805b : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b151bfbb2508476e8e21a08c7b2404c0 : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e8ed50d93ee4b6e97744508072714a1 : SUCCESS in 1h 24m 40s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f7e281487fa4bf5867b1f359331c48a : SUCCESS in 2h 30m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cc573bb93204018b9e2365e7120f2f6 : SUCCESS in 1h 59m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55ae7f71b38450ca57774f75fc702b4 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/def51b41d8cb45409b2e4e982588e132 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a1427fa2429fbc19e61bb59abe1419cd4dacffbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 10:00:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af22cf7ac8bcb66a5d5f070fb953afd68d24e599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 11:41:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4186899bcc34065baeaa80b21d8070f : SUCCESS in 6m 17s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/650f383acd884f3b9f8fb9881bc6603b : SUCCESS in 8m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4b87d1ba768a4c59a844223eeb7cb14c : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/23ab89962f0345e59a2e164a596eaf59 : SUCCESS in 1h 33m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0ec64c1a56964963a6740129eff7fd15 : SUCCESS in 1h 22m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb190b9250454e3eabba866918fdb39d : FAILURE in 1h 24m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09190796ce424beab035d82e933e703c : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3cab75f058f4fc1b0494bf62dbeb3b9 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"cc6eacf5d1bb6dde8b276cc7b6e69d58fa1d1c27","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-13 11:43:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"267d9fd4fdf10cf7936a3b768005108aeb073dc4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-13 12:10:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5c451d4a937c14ed452a2f7c18ca3f242fc36b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:24:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d9d8de6f0f749338aac96cfddce5c88 : SUCCESS in 3m 08s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1643ef5d69db46eaa5e5ed35d47f7b06 : SUCCESS in 9m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4c758805786346eabf94ce04f888f383 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca8923dc60d74f93b9dde923af50cc7d : SUCCESS in 1h 44m 48s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/70c5d3a5473b4abfb581f18e5af30faa : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dcc23dd75ed453d8054e5317205eb95 : SUCCESS in 1h 12m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b046a9c6307f4ff58c945253e43494ea : SUCCESS in 2h 35m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa81302705ec47fcaa0b165edd4aef23 : SUCCESS in 2h 20m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f786d20473a46f39038943d07ba7843 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/545f55fcc34448c3adfc04acd2316e67 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ce503041e856e54d5da726670b02a6069a34be8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:24:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d63d3b3654df94364056c977340da53c2347c190","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:02:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c8f3892cdddf9835a080e074646078c41957e219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:02:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/314159b3d2b94c9789085577bf444348 : SUCCESS in 3m 53s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/14ff7897d77a4d25bf4575afb3a28382 : SUCCESS in 7m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/42cd8595cd4b42d2a2fb9ba61f4c2baf : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d252b2ded2fb4e318cb897d74e71c129 : SUCCESS in 1h 32m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/064712f90ebc47fd8226fc79bfeae24c : SUCCESS in 1h 14m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbfa70e031343d7be5fb5f2a5ee95df : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88aaa12d61a042e09933fdd545e1b033 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"09e6c3423281d93d339cb6ed906eabd058669e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:04:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac7b2ab070341c683a8e691f085de52 : SUCCESS in 2m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c4a319e451a4770aa5211945696f299 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14bae8367056f2ce760a87f4ae1d26ce49543de7","revisions":{"0df9c9a8e8e45ca95a9c411d05885bb730c3afb9":{"kind":"REWORK","_number":1,"created":"2020-05-12 17:52:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/727332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/32/727332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/1"}}},"commit":{"parents":[{"commit":"dade9a647eb4ccbda3b716a233610225f0db5cb3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/dade9a647eb4ccbda3b716a233610225f0db5cb3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:52:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:52:51.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I88fc0d606ca0bffd4b2e44e3b2202955c38d4b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0df9c9a8e8e45ca95a9c411d05885bb730c3afb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0df9c9a8e8e45ca95a9c411d05885bb730c3afb9"}]},"branch":"refs/heads/master"},"14bae8367056f2ce760a87f4ae1d26ce49543de7":{"kind":"REWORK","_number":2,"created":"2020-05-12 18:05:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/727332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/32/727332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/32/727332/2"}}},"commit":{"parents":[{"commit":"dade9a647eb4ccbda3b716a233610225f0db5cb3","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/dade9a647eb4ccbda3b716a233610225f0db5cb3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:05:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:05:16.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I88fc0d606ca0bffd4b2e44e3b2202955c38d4b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14bae8367056f2ce760a87f4ae1d26ce49543de7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14bae8367056f2ce760a87f4ae1d26ce49543de7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
