)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~732325","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fussuri~Ifd9f8229bb4707466092c33e2aa2b0e118fd0aeb","project":"openstack/openstack-ansible-os_trove","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ifd9f8229bb4707466092c33e2aa2b0e118fd0aeb","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-06-01 13:19:21.000000000","updated":"2020-06-01 18:29:52.000000000","submitted":"2020-06-01 18:28:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732325-1591036096048-15326c95","meta_rev_id":"dd73df60385ee0e5943cd2c03f089fc94140f067","_number":732325,"virtual_id_number":732325,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 18:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-06-01 18:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-06-01 18:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 18:19:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-01 18:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb4d0d6effa48a3ea69f211e9400aa874277e746","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-06-01 13:19:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bd4bbb39332e026ceb6515693894a0b323f00c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7a72652d3144d31a0c92cb84d34192a : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d17d4fde6ef45fbb79d0db58ea255ef : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e511d21d0104e98b6dde0356ae2a6ab : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"89af696a1af1f5234e7de6bd969299651ef09429","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-01 18:19:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33192a95aa8eaa8ddb418a04f5c63ee0e5f44ccf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a02b4d80580891b67f94020a1f79d40fcbf165e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:28:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d7269da9f74c496d9d9da910005a6a16 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b88addabfe243bd8ea123dbe55d3b2a : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4163d278f29f41749b6257d6f52a5fc2 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"af84b00f4bae253162303f448766d6b729cd9757","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:28:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a82089ee3b8ddd2cf511fc03725632efc12b40e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:29:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bf6c5cc2e14b28a14ca9e986687a81 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65c12c80579f4c2e97dfc42ffcef57a1 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c75aea70b718d3c8ababb604b9238dd858a66a07","revisions":{"c75aea70b718d3c8ababb604b9238dd858a66a07":{"kind":"REWORK","_number":1,"created":"2020-06-01 13:19:21.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/25/732325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/25/732325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/25/732325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/25/732325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/25/732325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/25/732325/1"}}},"commit":{"parents":[{"commit":"d8079f63a57d386ca4b45a48d5b6bbab637a170a","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d8079f63a57d386ca4b45a48d5b6bbab637a170a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:19:19.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:19:19.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ifd9f8229bb4707466092c33e2aa2b0e118fd0aeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/c75aea70b718d3c8ababb604b9238dd858a66a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/c75aea70b718d3c8ababb604b9238dd858a66a07"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
