)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~734291","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fussuri~I7f82e588e297d85a66c20500adc0b06da895717f","project":"openstack/openstack-ansible-os_trove","branch":"stable/ussuri","topic":"osa-collections-stable/ussuri","hashtags":[],"change_id":"I7f82e588e297d85a66c20500adc0b06da895717f","subject":"Use ansible openstack collection","status":"MERGED","created":"2020-06-08 21:01:43.000000000","updated":"2020-06-24 18:10:28.000000000","submitted":"2020-06-24 18:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734291-1593022122527-91b529dc","meta_rev_id":"941cfd4488a7a7dba62890a9618f608021929069","_number":734291,"virtual_id_number":734291,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 18:08:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-17 19:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-17 18:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-17 19:01:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-24 12:44:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 19:01:20.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-24 18:08:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"291a228559d9e92924925c99dc2c1bb1df63884c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 21:01:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"987af72c95a5ab9334790d8adaf182f11dcf7bec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 21:01:48.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"88133d4cf0b12de1e59d929723e4b038f5e9379b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:47:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/49997cc5f9fb4a9faf57503333b2caee : FAILURE in 9m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/158556967eb648e8b3db974f45d53255 : FAILURE in 32m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf53de585f0b4f0eb545ddddb465f3bf : FAILURE in 38m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8aea93da459d4fd0b24245558478e233 : FAILURE in 32m 25s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/407b44fa7360449d9e8b7d04de603c87 : SUCCESS in 2h 40m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/909f49a4ee58404b9eb2d67100fb7658 : SUCCESS in 2h 18m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95cb6469a36463ea5f639a4f9c890d7 : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8a580a4774e55da73d0107ea3fc1286eb4908a73","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 06:08:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b17fc9bb119d2ac5c34e1c9e638a664269a8745b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 08:15:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/61420e83c4c640eaaa93a13e49032f29 : FAILURE in 10m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ee9f97d1aefb4c87ae86223bd78df7d6 : FAILURE in 33m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0be9eda07b4f44ed9028cab892a6dfd4 : FAILURE in 36m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2191c9075bf04cdd90dcfcb09e2d65f8 : FAILURE in 27m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/36d3c35b7c264d70af0290d0cf4ea039 : FAILURE in 1h 47m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1cf6cdce0794b6e96dc923431d29d81 : FAILURE in 1h 17m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d1a60a76224acfb3a8c1165c1e00be : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":2},{"id":"a98be52f9588b8cfb447635fc7631cc012029358","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 12:53:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ef2bda49410e08523777a332d3f24421dc609ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 13:56:15.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"e1e0fcad2ebb7a246220157254d58362c4badb41","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 13:57:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"800737dda8ba2066694d0b365df9cf61dae6a700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:27:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/88e5487840da43f3b89c1959c2d730b1 : SUCCESS in 10m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d5e0f0d52570463ba6f6e60665a68f02 : SUCCESS in 1h 16m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/25b826596a0b4135944d4d9d3dee4c12 : SUCCESS in 1h 04m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11e1ea8c38ce4c6490aa39700f9193f4 : SUCCESS in 1h 02m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a3a6134b61f04a07896162933d025cf3 : SUCCESS in 1h 42m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30090cf787664642acd60e936a939bcd : SUCCESS in 2h 16m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325d129f404947969b51edfd79052ee2 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b0869006043983f092cd3d89001f95e3b57505cf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 18:22:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8e007f2fbcebed94218f9a9f7b6b43ecfc92397b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 19:01:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"520d972c9a323665d82984f5122eae2f3c0376fc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 12:44:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"73448bc92ee930ea21c1a16c5c3f7b1eee7da901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:44:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b652ae9a90548961a26a4b7f07a4cca835c76f4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:08:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/17d298d3b1dd4264837ae2f15cdaf2a4 : SUCCESS in 7m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/79b75fb18be2412686b7b054aaf8e6cb : SUCCESS in 1h 33m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b419d90abb5d4e01a432a30098cf9d96 : SUCCESS in 1h 42m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2db1f715e6a481392d53eb6bf675c0f : SUCCESS in 1h 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af09eeffc92e46a899d6624c9c03dbaf : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"471821bcb9f52e357f1ec86d368dbe5788ddabc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:08:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9d7d3c8aa347ed3a046d4a348e5d574b648bb88a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:10:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337167a38bab4b72b2d420190855fe4f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fdc36622e9c33a6b36b9862509a63a00331aad1","revisions":{"88809d6bcb0b1e3218c368022bbafa5fea666aa3":{"kind":"REWORK","_number":1,"created":"2020-06-08 21:01:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/734291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/91/734291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/1"}}},"commit":{"parents":[{"commit":"c75aea70b718d3c8ababb604b9238dd858a66a07","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/c75aea70b718d3c8ababb604b9238dd858a66a07"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:40:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-08 21:01:43.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: I7f82e588e297d85a66c20500adc0b06da895717f\n(cherry picked from commit e07ea77e5830074187c815b431d68f612722e643)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/88809d6bcb0b1e3218c368022bbafa5fea666aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/88809d6bcb0b1e3218c368022bbafa5fea666aa3"}]},"branch":"refs/heads/stable/ussuri"},"7fdc36622e9c33a6b36b9862509a63a00331aad1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-09 06:08:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/91/734291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/91/734291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/91/734291/2"}}},"commit":{"parents":[{"commit":"c75aea70b718d3c8ababb604b9238dd858a66a07","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/c75aea70b718d3c8ababb604b9238dd858a66a07"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:40:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-09 06:08:07.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/734419\nDepends-On: https://review.opendev.org/718362\nChange-Id: I7f82e588e297d85a66c20500adc0b06da895717f\n(cherry picked from commit e07ea77e5830074187c815b431d68f612722e643)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7fdc36622e9c33a6b36b9862509a63a00331aad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7fdc36622e9c33a6b36b9862509a63a00331aad1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
