)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~772052","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I760b42b9f2c80932e3286d8c686d10b90f3f03b5","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"reno_build","hashtags":[],"change_id":"I760b42b9f2c80932e3286d8c686d10b90f3f03b5","subject":"[reno] Stop publishing release notes","status":"MERGED","created":"2021-01-22 16:28:35.000000000","updated":"2021-04-06 19:40:17.000000000","submitted":"2021-04-06 19:36:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772052-reno_build","meta_rev_id":"19bdc2c7a16f919bb09031a7387930661b6bece8","_number":772052,"virtual_id_number":772052,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 19:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2021-04-06 13:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-02 13:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2021-04-06 13:15:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 20:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 20:48:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-04-02 13:33:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-06 13:15:50.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"}],"messages":[{"id":"e723d2ebe8a63b301b2a967b37bdb7e2fbec7b4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-22 16:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ff045ff317b10b320dd2f9735ed7fe0afb9c4c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-22 16:35:40.000000000","message":"Topic set to reno_build","accounts_in_message":[],"_revision_number":1},{"id":"53893f771a8437a37262c2e40fa2b568d8e753dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 20:23:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/659cb9edcabc4469a8b85dd30be3bde7 : SUCCESS in 9m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/644e73b224f64a79a8023ecd4aac2bb8 : FAILURE in 1h 11m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4e7475269160433fae1b4b29644affb0 : SUCCESS in 1h 29m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/845ead10d7154a29bce7e068fb39a75e : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/615e22e521bd4892998859b4e54fe280 : RETRY_LIMIT in 2m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8ea33e4660a493fb74aed4a49951401 : SUCCESS in 2h 04m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d1a54d298f0489aa936fcf395ad9b95 : SUCCESS in 1h 54m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89df56dbb8e14c2584291b0dd5dc014f : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5fdcea4d68b6719181f30341a3b95e72b7304b4e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-17 20:48:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"57093b7b25b9b5ae56d031c0524ab8c6fb21cfc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 23:39:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3c0f3fdbcb084eea9331f663ef4691d3 : SUCCESS in 11m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/55fb5573faa74889b3d4fb24eec3300a : SUCCESS in 1h 18m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d43109988d964c71bf06ea057727d907 : FAILURE in 1h 15m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b5a3282518e4baf9f0455373ef71f8a : FAILURE in 1h 06m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a053265a679454abd897ecea8549d3c : SUCCESS in 1h 16m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e4d9ecf1a5db4da8980d412691294e15 : FAILURE in 1h 16m 18s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5ebb4a8923c472a9deaba19fb4ff156 : FAILURE in 1h 28m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a6bb60ad21456d82f95631ab0d558a : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3f8bc6b0cabbcbaf40ade46f982fc3a0d2acaba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-25 10:20:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"17c095f2ace7cd85bb1a720b9882915bd471d092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 12:36:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a3a5fb42abd743c9aea0e480c24532f1 : SUCCESS in 12m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/14a2e1bbfde743558e51d53ffeebafb4 : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/491bb37201684ba6861a323288c5e3af : SUCCESS in 1h 27m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62916fc038474d6ab0315813b8face78 : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd3508d8f27e422cb25d3984cdd522b8 : FAILURE in 1h 06m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3295686ed3fa4ebc9fff815acec58d37 : SUCCESS in 2h 03m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2a6018d49ec4c30a778d6be73ff2078 : FAILURE in 1h 06m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffd8dcb8f9c4332b77fccd054d8f85d : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a0ba89f2f02325451a96bd5ba507712463c14480","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-02 08:12:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba3393b93b0a49a0f98b3c87919568f2f587ada0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 10:13:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d557c51a019c4d7fbe036e71219f03f3 : SUCCESS in 9m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/80e1517dd98c4e758e165e9473be62d5 : SUCCESS in 1h 16m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90d8acb8ebc347a99f4083094ee80124 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa805815d9d44049866d84c745b66d61 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fc1132066134d6f90405cb98b8a7f57 : SUCCESS in 1h 11m 12s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/301d152a0da04be4b4b899755cf5eb33 : SUCCESS in 1h 55m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737911d6c8a04d2e93912b47a677962a : SUCCESS in 1h 48m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e188c4ed89874d4babbe7e4b1b47c9d1 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"79cdf05493e8d0177151379fd834966509ed2570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 10:13:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d557c51a019c4d7fbe036e71219f03f3 : SUCCESS in 9m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/80e1517dd98c4e758e165e9473be62d5 : SUCCESS in 1h 16m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/90d8acb8ebc347a99f4083094ee80124 : SUCCESS in 1h 19m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa805815d9d44049866d84c745b66d61 : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fc1132066134d6f90405cb98b8a7f57 : SUCCESS in 1h 11m 12s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/301d152a0da04be4b4b899755cf5eb33 : SUCCESS in 1h 55m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737911d6c8a04d2e93912b47a677962a : SUCCESS in 1h 48m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e188c4ed89874d4babbe7e4b1b47c9d1 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"dd3f7c6be2b4e88bb5bef38f7a953c928d37c0a6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-02 13:33:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2c5c1eb50e1d7e18a51ededca0a0cd84aee3e37f","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-04-06 13:15:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22d00d3e85be404ca0b551f38dd55b6488e9c145","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:17:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b0918a628225f648d1be6cf1ac07bb65b5180d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 15:02:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5a6d8777ac3647418506c04ff8a79878 : SUCCESS in 9m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/32dd783f75814c869fc7e79522d73a7c : SUCCESS in 1h 03m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/81d8e03519014ddda9c9549b9d51684f : POST_FAILURE in 1h 37m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5868e1c071b24650a051cf220ceb82c2 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e17652a545ea42f3b0da26db12c5a579 : SUCCESS in 1h 16m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12573ab7b4a444d7b07acc4b841593f1 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":2},{"id":"67f422b20dc5694defa52eaf96b19aad602eacd8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-06 16:09:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b8871abfbb2327e099c5db6293061df421608351","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:13:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f4790470972e4f41836bb861a0b8d6ab : SUCCESS in 10m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5b23026eb3da4472a64805c70bc3e3af : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/712b3ba18eee40f1a54176235ae45b1d : SUCCESS in 1h 19m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d266c5d099404bce8e0e1b4d4e2867c8 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e973040dbaf6413ea5283b6f1dd9c459 : SUCCESS in 1h 08m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d7090faf69eb4a939cc1ad6fd49b277b : SUCCESS in 1h 54m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/268c103335e8450889ae884487a0d614 : SUCCESS in 1h 36m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8156a629a1f43e8a07db3e73bfbda71 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"47519ff6f9c8b4f0be0dbe5e245753cad0afe612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 18:14:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"50592807a8c18bb2d8dc04e3a9c9c5690b8217b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:36:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/48bd8cfe8efe4ca395ec498cad0ff8aa : SUCCESS in 9m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f9bf528aa4864328895dc23134cb64cf : SUCCESS in 1h 11m 52s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0375d838a4ad4b329c1ab3f3ed4e393e : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e72bc05f735e4b06847fdb5fd2a30f61 : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/449aeecba1734ada9aa2d8713bbfda1b : SUCCESS in 1h 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2062390ebe2743d98d88c19951dfc429 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3e61f63ba75be26fefb67580afd19391096beffa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:36:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"19bdc2c7a16f919bb09031a7387930661b6bece8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:40:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2392a22a535746b5956eb3b021a9974c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ebe2de5e79b77a4d790fc15d9e74a788ce478e4","revisions":{"1e9c7699ecd0ccb16a93fe2d5bd46c8c2085f382":{"kind":"REWORK","_number":1,"created":"2021-01-22 16:28:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/772052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/52/772052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/1"}}},"commit":{"parents":[{"commit":"0629e28768b14ca976cceb4320add64d76420384","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0629e28768b14ca976cceb4320add64d76420384"}]}],"author":{"name":"dmitriy","email":"dmitriy@dmitriy-XPS-13-9380","date":"2021-01-22 16:28:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-22 16:28:30.000000000","tz":120},"subject":"[reno] Stop publishing release notes","message":"[reno] Stop publishing release notes\n\nSince we copy all release notes to the integrated repo there is not need\nin publishing release notes for each repository. We should only verify their\nvalidity and linting.\n\n\nChange-Id: I760b42b9f2c80932e3286d8c686d10b90f3f03b5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1e9c7699ecd0ccb16a93fe2d5bd46c8c2085f382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1e9c7699ecd0ccb16a93fe2d5bd46c8c2085f382"}]},"branch":"refs/heads/master"},"1ebe2de5e79b77a4d790fc15d9e74a788ce478e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-25 10:20:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/52/772052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/52/772052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/52/772052/2"}}},"commit":{"parents":[{"commit":"d14cd8cea43ff32a4e9213cc8eabca5b3704cb9d","subject":"Retry on creating trove network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d14cd8cea43ff32a4e9213cc8eabca5b3704cb9d"}]}],"author":{"name":"dmitriy","email":"dmitriy@dmitriy-XPS-13-9380","date":"2021-01-22 16:28:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 10:20:14.000000000","tz":0},"subject":"[reno] Stop publishing release notes","message":"[reno] Stop publishing release notes\n\nSince we copy all release notes to the integrated repo there is not need\nin publishing release notes for each repository. We should only verify their\nvalidity and linting.\n\n\nChange-Id: I760b42b9f2c80932e3286d8c686d10b90f3f03b5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1ebe2de5e79b77a4d790fc15d9e74a788ce478e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1ebe2de5e79b77a4d790fc15d9e74a788ce478e4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
