)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~782314","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I866fea573429ddcedf1547d9fa8a7caae448eca2","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I866fea573429ddcedf1547d9fa8a7caae448eca2","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2021-03-22 18:12:33.000000000","updated":"2021-04-06 14:03:22.000000000","submitted":"2021-04-06 14:01:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782314-policy-json-to-yaml","meta_rev_id":"5bff6fc91029e2780d8054b53c9b86bbbc40fe62","_number":782314,"virtual_id_number":782314,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 14:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2021-04-06 12:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-31 06:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2021-04-06 12:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 20:52:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 15:15:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-06 12:22:51.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"}],"messages":[{"id":"fd3e20e09eae67f5f1bb7178476d64684d686415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-22 18:12:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01a6ec17fbe18412c6eab06e0c5d97cb1e3c7bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:52:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d3fe5b3ceb8f44c493457008a71d7276 : SUCCESS in 8m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/832c5a83c2424abb8f5cc7b925e1c067 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2bb9dc2b570c47efa9a760a53d281875 : FAILURE in 1h 12m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9714fd5e29384df0ab2aca68ff41546b : FAILURE in 1h 08m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0a7d881e8f341358f22d831339c3800 : SUCCESS in 1h 16m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/92b7d4ea974d484da3bb244c2369960f : FAILURE in 1h 24m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77cf9b04057145c6a8e4dd65efe5064d : SUCCESS in 1h 18m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e5573f5cb641039ceb5804e36d6a6c : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a3c78b4a3c5db0f5ffa8643e7ef7bc38cd71b4b3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-23 08:22:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c8da9ab190ecdfce0717c3f53007f7344112863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 09:52:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4639806d08344f3fa344deeaed7c5132 : SUCCESS in 7m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9034ac8208b545189e6952e0d3505142 : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0bd6171214154c689bcc3550ebd2f6f2 : FAILURE in 59m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e11feb1b078b4617b0a6a89e5e07a113 : FAILURE in 1h 08m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b696e83e12e47ff8208eebed1795b44 : SUCCESS in 1h 08m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/10cfb28de04f465c81c40676b04872a5 : FAILURE in 1h 22m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee53646ff6474cdc95ea0bac778cda1c : FAILURE in 34m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f12a30d2fe485fb1d49ffa98d5f5bf : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e6a7de91c2d615af6988bab869d9da3c6e449812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-24 11:59:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b5c42b9132ab05517f59eaeb31409b4db42d6981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:09:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7abf8e2fb1ab47a2908e6c67c5850d36 : SUCCESS in 10m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c0bdc2f44f6d4865ba43acaeda3b3579 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/538691350b274bb49290319033d50f9e : FAILURE in 1h 16m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d786df26e724d039c15886967ae7b47 : FAILURE in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/027a986d53924c1bba2b860cdc1d16dd : SUCCESS in 1h 09m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a406539c9ae5438bb7367b212c80bebe : FAILURE in 1h 45m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf23f28c8f6c4fecae677680ac933f65 : SUCCESS in 1h 59m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29375d1b427a44a7bd1c55fbcf35ac32 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4d04ed6998392a562e0856e74e35d6ad221bc4f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-25 10:20:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ecc9bba646f7a36392cace7726f5a84804bf9d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 12:18:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2888c46d686b488b806e3fcd2302ef99 : SUCCESS in 9m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eb6c517335ef470baa0671ae0406a2f1 : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/85925e27dbfa4e3d9029a7bf42a1f7eb : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dc02e835ffb4adc8a813100b1b0783c : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8343c3cdd4e64a32b60672826f35f053 : SUCCESS in 1h 15m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2f548feed1ea4a49b9c85291fd4058fe : FAILURE in 1h 06m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95885098c59240869b56037dd363764e : SUCCESS in 1h 46m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a677dfeb7f43429c83edae1acd9fc34c : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1653370c3aaf3a66d34a12843374c644dc633c03","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-30 15:15:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"758f0483b1d4ac91d2f575ec655089b06c73d865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-30 16:54:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"164a5bdb427cbc0c153c16f279cc0ed57efd226b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 23:03:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f22c5540b8ea41c796e868a1afc4d640 : SUCCESS in 11m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ccf3b9fb20fe46f0bb483665eb8cff20 : SUCCESS in 1h 13m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8f3082e60b7142fa9e4843266d6b00a9 : SUCCESS in 1h 04m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43ce61eb4bc749d4843d62675563c68a : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0feb09c9d0874546be60d41ce93dbf5a : SUCCESS in 1h 16m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/393914b42c474ae9bac8bcdb3379cdb3 : SUCCESS in 1h 59m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aefbcf32a9284c67b915993508100a07 : SUCCESS in 1h 39m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d9249ef0334f8288104a431f8a2d97 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":4},{"id":"7cadbd0767e4042ff37d572e5a9a99bf43a91a2f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-31 06:18:38.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"25388928a8ce2eab4239489179be1ad8a2c9b740","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-04-06 12:22:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f962533fb00f8d9e248f99e30189a22ebe186bb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:23:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"869608d329b7ce003868aadb1e1fef1de3453d3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:01:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e8f39ca76add4f7aab7a6e420cb82deb : SUCCESS in 11m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9559d62709c54e34911007ec89e0ade1 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ce77e19611504cbda945686b40b9b175 : SUCCESS in 1h 27m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50b1d72387e4443eab71d7ebf2d5ac9d : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/581a36abd59a4666a85e8a853b122a4f : SUCCESS in 1h 14m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef2e5cae0b64a0ea4d068b4c959178d : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ab37f461e9058308f19412e4608d7465a3ab3ae4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:01:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5bff6fc91029e2780d8054b53c9b86bbbc40fe62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 14:03:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390ba318f7f847278004dfbcd565b43b : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eccf80bb7f02c46d382e4c5412539d3a618016a7","revisions":{"9e3624302c7cf3d08a0b155694d91651c2ab73e8":{"kind":"REWORK","_number":1,"created":"2021-03-22 18:12:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/782314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/14/782314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/1"}}},"commit":{"parents":[{"commit":"0650e1d2ead89ea71bdb33c3650a20aa76f020d6","subject":"Merge \"Remove references to unsupported operating systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0650e1d2ead89ea71bdb33c3650a20aa76f020d6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-22 18:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-22 18:12:23.000000000","tz":120},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I866fea573429ddcedf1547d9fa8a7caae448eca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9e3624302c7cf3d08a0b155694d91651c2ab73e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9e3624302c7cf3d08a0b155694d91651c2ab73e8"}]},"branch":"refs/heads/master"},"e39590384a1cc1c0ed603de4a7b30fbf3e33183c":{"kind":"REWORK","_number":2,"created":"2021-03-24 11:59:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/782314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/14/782314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/2"}}},"commit":{"parents":[{"commit":"0650e1d2ead89ea71bdb33c3650a20aa76f020d6","subject":"Merge \"Remove references to unsupported operating systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0650e1d2ead89ea71bdb33c3650a20aa76f020d6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-22 18:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-24 11:59:42.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I866fea573429ddcedf1547d9fa8a7caae448eca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e39590384a1cc1c0ed603de4a7b30fbf3e33183c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/e39590384a1cc1c0ed603de4a7b30fbf3e33183c"}]},"branch":"refs/heads/master"},"9ea2fdca91da72edd926475615447eb785ceaa4f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-25 10:20:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/782314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/14/782314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/3"}}},"commit":{"parents":[{"commit":"d14cd8cea43ff32a4e9213cc8eabca5b3704cb9d","subject":"Retry on creating trove network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d14cd8cea43ff32a4e9213cc8eabca5b3704cb9d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-22 18:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-25 10:20:55.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I866fea573429ddcedf1547d9fa8a7caae448eca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9ea2fdca91da72edd926475615447eb785ceaa4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9ea2fdca91da72edd926475615447eb785ceaa4f"}]},"branch":"refs/heads/master","description":"Rebase"},"eccf80bb7f02c46d382e4c5412539d3a618016a7":{"kind":"REWORK","_number":4,"created":"2021-03-30 16:54:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/14/782314/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/14/782314/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/14/782314/4"}}},"commit":{"parents":[{"commit":"d14cd8cea43ff32a4e9213cc8eabca5b3704cb9d","subject":"Retry on creating trove network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d14cd8cea43ff32a4e9213cc8eabca5b3704cb9d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-22 18:12:23.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-30 16:54:44.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I866fea573429ddcedf1547d9fa8a7caae448eca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/eccf80bb7f02c46d382e4c5412539d3a618016a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/eccf80bb7f02c46d382e4c5412539d3a618016a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
