)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~879987","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ib4f3ced7f2175e5187f8c16f88fbfbcb3a472d0b","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/ansible-collection-2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-12 19:54:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib4f3ced7f2175e5187f8c16f88fbfbcb3a472d0b","subject":"Adopt project_info fetch to collection 2.0","status":"MERGED","created":"2023-04-10 16:30:55.000000000","updated":"2023-04-12 19:55:22.000000000","submitted":"2023-04-12 19:54:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879987-osa/ansible-collection-2","meta_rev_id":"2ed394b6e33e909222ddd1608433f5002e7b488b","_number":879987,"virtual_id_number":879987,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 19:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-04-12 18:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-11 07:40:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-04-12 18:02:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-10 18:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 07:40:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-12 18:02:32.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"eea03955bbac4324c8902ce6666719294857206e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 16:30:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f53be876e5bcee3d9d9dffbb98c33c606654fc45","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-10 16:31:03.000000000","message":"Topic set to osa/ansible-collection-2","accounts_in_message":[],"_revision_number":1},{"id":"b984825772a2ea9e4c4d6160fc87249a47cd4a4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 18:43:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9184424938174bc6b9697da0ff371b36\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fc88f4f20964e8b853269bb092fc556 : SUCCESS in 10m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9014cd61f28a4943b5d5bef7df82ded9 : SUCCESS in 1h 15m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/daf4366c00f3489d9214c4b833558d26 : SUCCESS in 1h 26m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/520b1b4eeadd47b695f1954654f764bd : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19e05c948fca4971bfa10154c5736818 : SUCCESS in 1h 27m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0318e47a25754fc7ba3867ac62fd1d89 : SUCCESS in 1h 34m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5ba6da1fcab84f96ad559cfb4c62a15d : SUCCESS in 1h 58m 16s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e87e6ec9dce6418eb24a7239f44074e3 : SUCCESS in 2h 09m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ddffb9266941c79e3ad2d276768362 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5971210c8bb17bc59a923d1ffab43e7183927d01","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-11 07:40:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86a08e8e39fa571ea618442dad7e67e255c309a9","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-04-12 18:02:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b382b160948475751b21363108bcf2d2cb0a253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 18:04:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0e5961bff7fdb59fab3424df983b7425eb07ba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:54:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a1b28ce6824eb698f52e5acee89495\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f33a4e28196048f6821e4238b5cf4f95 : SUCCESS in 13m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/aa8cd07989594e3381e5b96152105b80 : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3cee1e4d969e4984a7dd3d2ec6d9669b : SUCCESS in 1h 43m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f4818e242724d0da6f41ed384d13a83 : SUCCESS in 1h 35m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08327788bc9d487c893cf53e96c2a1b1 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a94200f4b851485859aa2706753563838702dff0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:54:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ed394b6e33e909222ddd1608433f5002e7b488b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:55:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61cce2a51f284f7e854e8bcb24168756\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba61d82924049e6be7795da6b1697f1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b326d50c7d8b645c4e3db93cc2e14baa1863c3c","revisions":{"7b326d50c7d8b645c4e3db93cc2e14baa1863c3c":{"kind":"REWORK","_number":1,"created":"2023-04-10 16:30:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/87/879987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/87/879987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/879987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/879987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/879987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/87/879987/1"}}},"commit":{"parents":[{"commit":"fde458164b4c4536466f7609991324a3f051d35d","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fde458164b4c4536466f7609991324a3f051d35d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 16:30:47.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-10 16:30:47.000000000","tz":120},"subject":"Adopt project_info fetch to collection 2.0","message":"Adopt project_info fetch to collection 2.0\n\nWith ansible-collection version 2.0 return of project_info module\nhas changed. We need to adopt usage of module return to the new format.\n\nChange-Id: Ib4f3ced7f2175e5187f8c16f88fbfbcb3a472d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7b326d50c7d8b645c4e3db93cc2e14baa1863c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7b326d50c7d8b645c4e3db93cc2e14baa1863c3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
