)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~882053","triplet_id":"openstack%2Fopenstack-ansible-os_trove~stable%2Fzed~I53f50f5c7e56bdadd0456bfdeee1900a5fc91a89","project":"openstack/openstack-ansible-os_trove","branch":"stable/zed","topic":"osa/pki-stable/zed","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-05-04 10:44:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I53f50f5c7e56bdadd0456bfdeee1900a5fc91a89","subject":"Add variables for rabbitmq ssl configuration","status":"MERGED","created":"2023-05-02 16:27:51.000000000","updated":"2023-05-04 12:19:09.000000000","submitted":"2023-05-04 12:18:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882053-osa/pki-stable/zed","cherry_pick_of_change":880760,"cherry_pick_of_patch_set":1,"meta_rev_id":"e48415bcf6bc72ab923bdce038dce5796c2f33a5","_number":882053,"virtual_id_number":882053,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-04 12:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-05-04 10:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-03 13:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-05-04 10:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 18:46:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-03 13:38:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"f066fea849594ca7b1465d17e204965feefe64d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-02 16:27:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6615904f1726a3b0d91c23f8485f67160569c062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 18:46:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309c5b4d767946348995e1d3a2cd10b7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fa2547a852e49e78bc04441b5841c45 : SUCCESS in 15m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b01b5074417343a58441bedad4a7dc87 : SUCCESS in 1h 16m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad59d2fa6b884a059bb93fb987be8304 : SUCCESS in 1h 31m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2ef74280ee1469e85ca53843ed45db9 : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b55eefe9fe624d5d87741f953370b6ec : SUCCESS in 1h 29m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b651ed0659f414d8e63fb65b19ec129 : SUCCESS in 1h 06m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d9e47da08c10499e9b55b793a35262c6 : SUCCESS in 1h 54m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f68a52f7429490aa3fae8501a1cc9a5 : SUCCESS in 2h 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd133754e0074e9f98e749220aff631b : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b398b82cf446375fe699591564868915df1ba8d8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-05-03 13:38:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19a3471d5173bdb0dfb19bbb0903a1f8fb757c7c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-05-04 10:44:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02dc2e6e5546d51167cfa7c6aef6bd04807e00d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:46:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dfe8c3cccc635391f7222b446acc2c4c1342805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:18:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc90cad8a58845a59c2d8916f72826a3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84dcecc3aece4eb98cdd9d398ae49dbd : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/1df6028baf3547d2b6a0d6c859f65b01 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/883a12af38e54fcba87725ea99420268 : SUCCESS in 1h 22m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dec8641a2bfa4ab4a6da74e7e2077bb2 : SUCCESS in 1h 30m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd27e1c1b7c4d169c82cb7e44b66aa6 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c501031abd99424f21c3883da50fa75f510b80f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:18:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e48415bcf6bc72ab923bdce038dce5796c2f33a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:19:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3addd6bc1324fb8b40713f0a4c7e703\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a6062f6cd24b20bd4f54f1c5440618 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b45d81b1ff081476da8556a524a8461ca6314b06","revisions":{"b45d81b1ff081476da8556a524a8461ca6314b06":{"kind":"REWORK","_number":1,"created":"2023-05-02 16:27:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/882053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/53/882053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/53/882053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/53/882053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/53/882053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/53/882053/1"}}},"commit":{"parents":[{"commit":"d8464db15af63cd6422ed1d2da271cb40e4d3b96","subject":"Update .gitreview for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d8464db15af63cd6422ed1d2da271cb40e4d3b96"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-04-18 15:25:26.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-05-02 16:27:51.000000000","tz":0},"subject":"Add variables for rabbitmq ssl configuration","message":"Add variables for rabbitmq ssl configuration\n\nChange-Id: I53f50f5c7e56bdadd0456bfdeee1900a5fc91a89\n(cherry picked from commit 189993434526c315d4dfb7e9250bc0a4481e5fb7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/b45d81b1ff081476da8556a524a8461ca6314b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/b45d81b1ff081476da8556a524a8461ca6314b06"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
