)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~917047","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ib68778a8cb8535d7400be04f02d332ba0344d20e","project":"openstack/openstack-ansible-os_trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-03 20:19:23.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-05-02 20:31:33.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Ib68778a8cb8535d7400be04f02d332ba0344d20e","subject":"Add independent set of variables for guestagent configuration","status":"MERGED","created":"2024-04-25 13:45:52.000000000","updated":"2024-05-03 20:20:21.000000000","submitted":"2024-05-03 20:19:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917047","meta_rev_id":"16fa7a5db955529310806869134c0b5c8a198bc4","_number":917047,"virtual_id_number":917047,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 20:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2024-05-03 18:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-02 20:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2024-05-03 18:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 14:12:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-04-25 18:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-03 18:25:29.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"81bf74136f49b0312acd9542b693fd302ebec72a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-25 13:45:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b575a8c9e05fff262fc77fcb5ad15423f2a5b16","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-04-25 14:12:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fbb6bb1573666dbdfdaf81ef8fe96172d30af715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-25 15:36:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"cadec8099b6855727ef08bba34d29a6359e53ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 18:28:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53fafde8915f4c029c3463e62b1791c9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6434ba1987d748a9a912922b4b42c3a5 : SUCCESS in 12m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a42547c7406d478f8da1fd9da52a730c : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a21d640767c14d0799cabc794d546947 : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/706ae674d610404c9c9a016debf3ad56 : SUCCESS in 1h 30m 21s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c6e052957344f7fb7dc920c748e7de1 : SUCCESS in 1h 30m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/712d8f4856c44bb3a32d4b53ba60d988 : SUCCESS in 1h 08m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ee90daf14e74ac28ad3dd46031daad7 : SUCCESS in 1h 31m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87c31556afd3453bb9adecbfaa26d1c1 : SUCCESS in 1h 14m 14s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e64543f35bf4f23a966ed854e932a9f : SUCCESS in 1h 09m 53s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b67f8d373284620a94773109b1de339 : SUCCESS in 2h 15m 26s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64aac951df0c43138bf015d7a9c99b3d : SUCCESS in 2h 44m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2c7e0accd6940069b6c1dafa63a81e4 : SUCCESS in 1h 35m 44s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05a0518a9cc24cb188153e7d6f027725 : SUCCESS in 2h 37m 57s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b650a177cb14463cb8f338d3a85b0090 : FAILURE in 1h 33m 31s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/142cbb90985f4a63bfb1ac5d25547540 : SUCCESS in 2h 02m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26dc47ceae2344ab9b057f03cfc9af2b : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff4658216ae4414dac6231243eb21c15 : FAILURE in 2m 39s","accounts_in_message":[],"_revision_number":2},{"id":"53206621b657f9ac50caa6a964cb438d98966e13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-26 10:04:08.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a4f86adccbf11f6a265f8cec18eb38cb94548bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 13:01:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07b32b552cd6479daa9fa597b0d746b2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18534fd5d2aa4cc39169d874a8ff70d4 : SUCCESS in 12m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ae98c2f327aa4f2186db9a6b2983985e : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7a7b509ed3114b70bfaa3e3bd617c5af : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddec0e18c87c4fdaa57d04f2e26d4513 : SUCCESS in 1h 29m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5cecf1c3375845d897db97c271322cce : FAILURE in 1h 19m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/911796fbcdca42df9feb4ddbf8380349 : SUCCESS in 1h 28m 56s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/52221432064c448996dc8e5749750206 : FAILURE in 1h 08m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b7f1a8188c04565bd53308f6ad630c8 : SUCCESS in 1h 28m 30s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8ce3b1e5c4d4e918548bcb4e8e34def : SUCCESS in 1h 21m 29s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d3ee339dd4d4de3a36433b70698e991 : FAILURE in 57m 16s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9c49bc016f8472291f490288825fa35 : FAILURE in 1h 16m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fdca9fa3cdb4ef48ed87bd5ce2686f5 : SUCCESS in 2h 13m 21s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b91ce5141b2647f4a335175bca61fec1 : SUCCESS in 2h 48m 15s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8c3b95e7d4c743cfbc51cbd9b1850202 : FAILURE in 59m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d586645d104c441597e3f75fc00649d5 : SUCCESS in 2h 05m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25344e703dd448c0aaeb9f9e8a8ef17b : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1a20cf8c9bb4d74abfffd37298b1f70 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2656d724eb9c87a2ea768e3d1f34f82f3bd73a9e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 09:00:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"624fe55aa47328a6393a3e322fe258434a414f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 12:00:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b007cd48fb54e38a08b77d6039fcfff\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/037557cc93744d89b630231479e141a2 : SUCCESS in 16m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4cbf5ae1839e48139a0a984a270d6436 : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/c15d1e6742724df1b3c1f06aef509cbe : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b14e99e566c4734b9b4dbfce9e435f6 : SUCCESS in 1h 32m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/627bc5b23d4346c59b7cc184169436ac : SUCCESS in 1h 21m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0361cafdc49644b4b0d8c10b7e0a4380 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8973566e22214a56875af9c96b61dd5b : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/259e0604eb674887bd5aa9d02354771c : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/815d99d33d7a427bb14ecda62fd8b04b : SUCCESS in 1h 14m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c40c1d91cdbb465194b3164471f3ef61 : SUCCESS in 2h 18m 57s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a02b66230bdd43ff9d26cf8cdb030782 : SUCCESS in 2h 34m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3070046c24bd4b33ba4bc831eea9e613 : SUCCESS in 2h 09m 36s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc5ef679222c452fb635bba8d9056aad : SUCCESS in 2h 41m 43s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/008bd9e20f1a4a1d9462a1631116dbb6 : SUCCESS in 2h 30m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcc065e148c94c5faf35545b80e79168 : SUCCESS in 2h 09m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e47d266ec24c0e95614b9955cdad33 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90783f71599a4c36b3261c9e2e425136 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":3},{"id":"bdc792fcbce75a343500f5edc7a47cd7e318fbb0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-02 20:31:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"83fc8a269129ce36972cb2c7d1f1df2a723539bd","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2024-05-03 18:25:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7435d002154ea6e98b818563026e83aafcca96bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:27:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9de1b1afe802c2c72954af560d36efefa9e45764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:19:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd34020bbebd4751978e4f34d69b2ac1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63da558138584a409ce5babe806620a8 : SUCCESS in 14m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dd78b0a3bd244d54b17ba8e41da8a1f1 : SUCCESS in 1h 14m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/180b801f9f774509980a9f3fbc803127 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/666aca887cd7490087cff1631e33a233 : SUCCESS in 1h 32m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a9fc129bc6743f4a76e36d40cd99cca : SUCCESS in 1h 24m 40s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e465359df044aa3929753e0ec806d71 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d647ebd40d9c48ce9a45f8c7c9d91471 : SUCCESS in 1h 43m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4633e6953b60456bb0f307c1b879dc00 : SUCCESS in 1h 19m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5ab117e4af47c0af4dc6d12b0e3815 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e067132ead04c3b9dd268e3b2f22878 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":3},{"id":"78534f06ef0125fd8b9f854a11971cf7a0b9f87b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:19:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"16fa7a5db955529310806869134c0b5c8a198bc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 20:20:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/054b25eafd2f462dbeca4ac3cf12e190\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b904710c19f74b14a3bae26a175c03e0 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0b4ddeeceaa086e9d23ae47a63ed07709c4b4b6e","revisions":{"2e22dc52caa705560adce3ac6e6b35c26e169009":{"kind":"REWORK","_number":1,"created":"2024-04-25 13:45:52.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/917047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/47/917047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/1"}}},"commit":{"parents":[{"commit":"110feb6128df34070872cb4f2fea6fab1cdc0496","subject":"Fix RabbitMQ vhost defenition for guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/110feb6128df34070872cb4f2fea6fab1cdc0496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-25 13:45:38.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-25 13:45:38.000000000","tz":120},"subject":"Add independent set of variables for guestagent configuration","message":"Add independent set of variables for guestagent configuration\n\nAs it\u0027s recommended to have an independent RabbitMQ cluster for Trove\nGuestagent, we need to make it possible to easily use also different\nuser/password set for it.\n\nIt\u0027s also important to control quorum queues independently, as migration\nbetween quorum queues and classic queues for guestagent cluster\nis quite cumbersome and potentially should be avoided as might\nbring severe disturbances.\n\nChange-Id: Ib68778a8cb8535d7400be04f02d332ba0344d20e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2e22dc52caa705560adce3ac6e6b35c26e169009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2e22dc52caa705560adce3ac6e6b35c26e169009"}]},"branch":"refs/heads/master"},"3c7a67c60b6ceeab0a8fbda9d20bc8451aee7b15":{"kind":"REWORK","_number":2,"created":"2024-04-25 15:36:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/917047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/47/917047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/2"}}},"commit":{"parents":[{"commit":"110feb6128df34070872cb4f2fea6fab1cdc0496","subject":"Fix RabbitMQ vhost defenition for guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/110feb6128df34070872cb4f2fea6fab1cdc0496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-25 13:45:38.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-25 15:35:19.000000000","tz":120},"subject":"Add independent set of variables for guestagent configuration","message":"Add independent set of variables for guestagent configuration\n\nAs it\u0027s recommended to have an independent RabbitMQ cluster for Trove\nGuestagent, we need to make it possible to easily use also different\nuser/password set for it.\n\nIt\u0027s also important to control quorum queues independently, as migration\nbetween quorum queues and classic queues for guestagent cluster\nis quite cumbersome and potentially should be avoided as might\nbring severe disturbances.\n\nChange-Id: Ib68778a8cb8535d7400be04f02d332ba0344d20e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3c7a67c60b6ceeab0a8fbda9d20bc8451aee7b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3c7a67c60b6ceeab0a8fbda9d20bc8451aee7b15"}]},"branch":"refs/heads/master"},"0b4ddeeceaa086e9d23ae47a63ed07709c4b4b6e":{"kind":"REWORK","_number":3,"created":"2024-04-26 10:04:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/917047/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/47/917047/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/47/917047/3"}}},"commit":{"parents":[{"commit":"110feb6128df34070872cb4f2fea6fab1cdc0496","subject":"Fix RabbitMQ vhost defenition for guestagent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/110feb6128df34070872cb4f2fea6fab1cdc0496"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-04-25 13:45:38.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-04-26 10:04:07.000000000","tz":0},"subject":"Add independent set of variables for guestagent configuration","message":"Add independent set of variables for guestagent configuration\n\nAs it\u0027s recommended to have an independent RabbitMQ cluster for Trove\nGuestagent, we need to make it possible to easily use also different\nuser/password set for it.\n\nIt\u0027s also important to control quorum queues independently, as migration\nbetween quorum queues and classic queues for guestagent cluster\nis quite cumbersome and potentially should be avoided as might\nbring severe disturbances.\n\nChange-Id: Ib68778a8cb8535d7400be04f02d332ba0344d20e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0b4ddeeceaa086e9d23ae47a63ed07709c4b4b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/0b4ddeeceaa086e9d23ae47a63ed07709c4b4b6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
