)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~917997","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Icd659cdf4c817d100668ab9c5639288b63078c38","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-05-14 15:36:32.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-14 15:36:32.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-05-13 08:58:13.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"Icd659cdf4c817d100668ab9c5639288b63078c38","subject":"Implement variables to address oslo.messaging improvements","status":"MERGED","created":"2024-05-02 17:34:46.000000000","updated":"2024-05-14 15:37:48.000000000","submitted":"2024-05-14 15:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917997-osa/messaging_improvements","meta_rev_id":"a138eaa16627557bcf535f529e123f3759a61cb5","_number":917997,"virtual_id_number":917997,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 15:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 13:50:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-14 13:50:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 20:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 13:12:05.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-08 06:53:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"383b7bdbc0adc6718538d6bfa2e95e071ae7f23f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-02 17:34:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"725a809adc561f2c3353bd12626b6e0241495aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 20:23:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c190134023344a97828a4d41201bdec8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe250584c41e45b5bb34a72c1ff9bf68 : SUCCESS in 11m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9f9c0755242d44458ea11b98b857f1cf : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/dae1de7a4b5c43d795c485264ad027da : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/76dafce023534e2097bc48c71bea3368 : SUCCESS in 1h 18m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf7c868d3ab5477da32df7f80186def5 : SUCCESS in 1h 33m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/584a0bb94e3d4b18af82cb6e4aadff4b : SUCCESS in 1h 14m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ec728550d95d4ef5b8cfd9230b747b85 : SUCCESS in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf3008a519a2475a9acaae8662951b9f : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61f39dadb5c5431aa4bb37193e22681f : SUCCESS in 1h 22m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/322627cb0b2b4ecfa39c2b1ed6c3e4ef : SUCCESS in 2h 02m 40s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5bad0ffbde4849cf807ca15e365fb299 : SUCCESS in 2h 30m 35s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc2c86609ab14a688a221299369bf6b6 : SUCCESS in 2h 00m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc71bb0865ab4a86a5ce11ef01fce821 : SUCCESS in 2h 14m 09s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f245f0c4b41d45cdadd0672de2a73afa : SUCCESS in 1h 47m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b06f40d3e4a04073aae829d332a09326 : SUCCESS in 1h 47m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb6c180b0384346b0a48f054601ab51 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3703201313630f409fbed9e2e03aab032e6e2bca","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-07 13:12:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91c718ac5c6a87c5472701976e2097c796ab3fdb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-08 06:53:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ca13a68b127469e41224d2fed6a84521bcadcd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-11 09:05:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"212ffd76e842006c774ba4ea83f85ef149427f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 12:01:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f13661fdb3847b387bdad6a1a9a005d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8055135485245e49f47e4978f6308da : SUCCESS in 12m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0eb3ce4598ae4f9299f364123ec22f60 : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/80593473872b47aab3ca2347238f65fb : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd0d609b8f2e4e21b5ca6b1bf353e4ef : SUCCESS in 1h 20m 03s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/be009e9ae7c84995b6599f2a57b4b6b1 : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/672d18e37cab499ba77ffb13fd359e71 : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/327dafba8bae41eebb12032a7218cb2f : SUCCESS in 1h 34m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2fb891565b146b1825eaf0a9ab87d60 : SUCCESS in 1h 19m 53s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da18ffc9d72d46dd96632d5f89b52197 : SUCCESS in 1h 23m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba58a5917f1c463dabb19e1673e24773 : SUCCESS in 1h 56m 59s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c94e5b708a6849408bafb0f4ae47d8c1 : SUCCESS in 2h 41m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84d662dcb170449daef04fb9e881ca18 : SUCCESS in 1h 34m 08s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d2dff17157e47c188eb266de86543f2 : SUCCESS in 2h 32m 05s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1163bafec6f44ce790d50940fa518497 : SUCCESS in 2h 02m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa3fcbbab1a448f383e3a18284e5dbe4 : SUCCESS in 1h 37m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f87a044c77214390818e3fa3f3e0b6b2 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"cd5e111b099480084dc793a4edb9707265fc9ef3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-13 08:58:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f898d54ffd2099dbdf2c6e26a958c63cb57a16f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 09:18:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e010ec97c73b00e84716b12796ce4fe602fba190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 12:03:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7a28a5ab804888869356a46d24c939\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1d6a97a63bb4d12b9a61c75dcc56603 : SUCCESS in 12m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/78b3c9ea96ef4982bf975e7bc1132a73 : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d614df12c1e0442ea60391dced392edd : SUCCESS in 1h 20m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff5fd0f8f94e40098b953a95c6083924 : SUCCESS in 1h 21m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2fe6601b824643269d30481944035411 : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fcc600ce45e4c239dc584aec09dbc72 : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a6b982841934b9a89abf5e43c90d857 : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f59385d320f48dca0aa8a10976f4131 : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b296f9317efe412097a37180537ea401 : SUCCESS in 1h 22m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/29f8d0899b4c482c9f773a8c64ca1c0e : SUCCESS in 2h 27m 18s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c638d591f6be41b0a6f21c4c7d2e69dc : SUCCESS in 2h 38m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f26a8724150d493d9458f69eea1af92f : SUCCESS in 1h 47m 43s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e20a1012d5e4bfab6062652ea7bac03 : SUCCESS in 2h 14m 41s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ff1098341f5445d9f03d1a7e30ffd47 : SUCCESS in 2h 03m 19s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1667f09a2d54fee80209e20c989422f : SUCCESS in 1h 33m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c5aea0fa99940ff857852e88c2ab173 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"60a6ae963765517049f2ec5ce2743965de5db107","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-14 13:50:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0e54128c7fd2f746dd70fe325d3f0706918c0779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:52:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4eef1f0dbfcb761fe3899238b3d8d549d32801c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:36:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b28c710521cc4c12be9cee3b36cb256f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f49add511844666817de5173e53f292 : SUCCESS in 13m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a2c225b4775e46099f4981b35c2e9db5 : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/58d41e5c56764b578ba2ac567da1e455 : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d725292569f84adeb6b34be0c12c75b8 : SUCCESS in 1h 26m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67979f0aacaa42c5acee5bc6589301aa : SUCCESS in 1h 30m 37s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9583bbf673a4a2ab14f14c87ae60703 : SUCCESS in 1h 35m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae569833b4f54137bd1acb58c29fd422 : SUCCESS in 1h 28m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e84f49ba3e84d54b80e71094cca3e0f : SUCCESS in 1h 20m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ca5dd03a2e460783156d2f764511f1 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"0db3cc28ce23d1a6623e09825cf40c2c1444a6f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:36:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a138eaa16627557bcf535f529e123f3759a61cb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:37:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/556ec7861a3e450bab51ee6c2a891f26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48805fe4c0374445aedff9b8a37bfe59 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"41868739632a59f0bd0410e69744b21e3916efda","revisions":{"7aa56eeb75211e925d9990406c2d593b6eef2204":{"kind":"REWORK","_number":1,"created":"2024-05-02 17:34:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/917997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/97/917997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/1"}}},"commit":{"parents":[{"commit":"c96668e78844ba3fbffbf8b041126c19f4d642da","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/c96668e78844ba3fbffbf8b041126c19f4d642da"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 17:34:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 17:34:34.000000000","tz":120},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: Icd659cdf4c817d100668ab9c5639288b63078c38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7aa56eeb75211e925d9990406c2d593b6eef2204"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7aa56eeb75211e925d9990406c2d593b6eef2204"}]},"branch":"refs/heads/master"},"5cf6094752d74f5af117339b77326e78b377b2bd":{"kind":"REWORK","_number":2,"created":"2024-05-11 09:05:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/917997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/97/917997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/2"}}},"commit":{"parents":[{"commit":"c96668e78844ba3fbffbf8b041126c19f4d642da","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/c96668e78844ba3fbffbf8b041126c19f4d642da"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 17:34:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-11 09:05:36.000000000","tz":120},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: Icd659cdf4c817d100668ab9c5639288b63078c38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/5cf6094752d74f5af117339b77326e78b377b2bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/5cf6094752d74f5af117339b77326e78b377b2bd"}]},"branch":"refs/heads/master"},"41868739632a59f0bd0410e69744b21e3916efda":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-13 09:18:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/97/917997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/97/917997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/97/917997/3"}}},"commit":{"parents":[{"commit":"3a8a9da7b958a9f37a68478d8f4ec026c6abc1a5","subject":"Add variable to globally control notifications enablement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3a8a9da7b958a9f37a68478d8f4ec026c6abc1a5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-02 17:34:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-13 09:18:05.000000000","tz":0},"subject":"Implement variables to address oslo.messaging improvements","message":"Implement variables to address oslo.messaging improvements\n\nDuring last release cycle oslo.messaging has landed [1] series of extremely\nuseful changes that are designed to implement modern messaging\ntechniques for rabbitmq quorum queues.\n\nSince these changes are breaking and require queues being re-created,\nit makes total sense to align these with migration to quorum queues by default.\n\n[1] https://review.opendev.org/q/topic:%22bug-2031497%22\n\nChange-Id: Icd659cdf4c817d100668ab9c5639288b63078c38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/41868739632a59f0bd0410e69744b21e3916efda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/41868739632a59f0bd0410e69744b21e3916efda"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
