)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~918103","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ib0126a0ac70aa613296a8d6e1ca61b34e22b02c2","project":"openstack/openstack-ansible-os_trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-16 06:03:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0126a0ac70aa613296a8d6e1ca61b34e22b02c2","subject":"Manage trove images through openstack_resources role","status":"MERGED","created":"2024-05-03 16:47:43.000000000","updated":"2024-05-16 06:04:20.000000000","submitted":"2024-05-16 06:03:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918103","meta_rev_id":"113826f6e25060091300f7e0b802ba75be13d820","_number":918103,"virtual_id_number":918103,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-16 06:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2024-05-16 04:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-15 21:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":1,"date":"2024-05-16 04:09:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 19:47:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 21:57:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-05-16 04:09:34.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"}],"messages":[{"id":"50833c68d1bfc1b4742363cd6895f1892abdf8fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-03 16:47:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e86db31bd027ee89b01281c48f0b10031af44a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 19:47:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678d71c441a44fcc9697439d76081ee8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eab38380159740b5a0b7aee86ea28bc8 : SUCCESS in 14m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dd8dd99b4a734bec88a7660c93650978 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ac4720969da24d81992c907bf5148030 : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad9e57bf1e8a44da83d37dbd294561dd : SUCCESS in 1h 31m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac6e9baaeea6496aa96aa69efd2270aa : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f919b62da9ee4eb48a397afbe8971bf0 : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a9874d8bd1f41a3b2c522872c06464b : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7456f4338984eab935829dd4dfc4846 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bffccce79d04ff0837a4372ac047b4e : SUCCESS in 1h 21m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ee17d2f1cbe4ded88715f19346eed4c : SUCCESS in 2h 20m 22s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cfbe36d71c2e4e0696ff500900910f17 : SUCCESS in 2h 49m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecabe577c8fd4239bf0da13d4f8140de : SUCCESS in 2h 03m 57s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/464fba99a27c4c6eb508e6ee68fe65b7 : SUCCESS in 2h 26m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4ef49435880f400a89e8fead8afd98e2 : SUCCESS in 2h 08m 55s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/075d1832dedf4da790a92beb44d1da52 : SUCCESS in 2h 16m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1118c38588764222bf1998c62975421e : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"70b32f5901448fbb48a6ebbda1ba2585afbc404b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-11 09:06:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c520505db87e409581b7fc0bb7073a2960df0455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 11:26:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78c5f1f140b245cb8f4bbfc44a1e4c33\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6f839e13a8a454caa2a50c43b64aefd : SUCCESS in 14m 01s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fc9647fac9264590a4da4507e8a56725 : FAILURE in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4c3048ae81d84f848a5010ed5f8e8017 : FAILURE in 1h 15m 31s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0aa741810cfd40acb7de9cbadf64e527 : FAILURE in 1h 12m 03s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/40270a4156644432ad21440ff4c4cbea : FAILURE in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81f345ac88e74529bd93383e29b53d22 : FAILURE in 1h 21m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/32fa4a9b5ff04a30a782ac3702a0eee9 : FAILURE in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2856d85dba174aeaa2515b699957667e : FAILURE in 1h 07m 56s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d952592d1a543e1969a208894aff0a9 : FAILURE in 1h 18m 48s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fabade1cb5244ef08bd52c614434b2bd : SUCCESS in 2h 16m 27s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a18762bcdcb4006a940bf58d1aadbf2 : SUCCESS in 2h 12m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb15358724ff4e33babdfe1d19b4a9c0 : SUCCESS in 2h 05m 18s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5de2a4d78254acf848070ecf9d2adad : SUCCESS in 2h 13m 57s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8d80b409fbfd4b6ca087f09e1c07eb1c : FAILURE in 1h 59m 09s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec2dcae1a1ab403f8ba6b7071adf70c2 : FAILURE in 2h 05m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74c9257e2a94e00aa78c1b70e065961 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5c3c523733e78817684ee0a994884c12c3bac5c8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-12 08:03:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d8562627d9960533a33161030c523a546ce426e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-12 10:43:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ebe2a3414c94a0c84127537d9e82c1d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a73f2960388f4071b673515bfedf1684 : SUCCESS in 12m 57s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7b570ef1442a4f2bb7c60f30b911e591 : FAILURE in 59m 24s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/28c7eca32e43424c8cb9a3adfe8d4791 : FAILURE in 1h 05m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed2cf9b076c74116978756d1486da4e3 : FAILURE in 1h 25m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0f2c96ce03614d69b9efc6f572dd6177 : FAILURE in 1h 29m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbd8edd513f44671abe6aea3c45bb238 : FAILURE in 1h 10m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8fba3b27b4f54172a8076c007d37bf6b : FAILURE in 1h 31m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8e5e240285a4c3080ead78b58ed7188 : FAILURE in 1h 22m 13s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edf4277b47094febaae25e54e5c8c06f : FAILURE in 1h 20m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81db9f016139400ca37a68b8b21ecf64 : SUCCESS in 2h 04m 56s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d38aa47c6ff446dd9b461a94a5e24050 : SUCCESS in 2h 33m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e30997aaafa44c9c84e95f8e110d9f70 : SUCCESS in 1h 59m 32s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8f2c1bb3d484b4095a2c77aec6c1b58 : SUCCESS in 2h 27m 48s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c385a5cb3bcd4312a41dacdd9a2be24f : FAILURE in 2h 07m 04s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ad7997484784c68a6a0e060e92fbdb5 : FAILURE in 2h 00m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c71e9eea055437b852a8de41aea493d : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6bf9759582268375e00a2b7a95b06e938bcfcdb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 09:45:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7387f6a2dc22442e7a90526c131f696d4b9ed630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 09:46:44.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/784379 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"e335c929dfdb21b75c1081bf07ae4f1f816018ac","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 09:47:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"177c9a2d0eb7a988ebf6ab0e70e757e4567d1ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 12:36:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90ed647cae49449ea36028f4688ce0d0\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62f774fb9b8e4fc183cebeb1868804fb : SUCCESS in 12m 06s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0cd2b7bc8a4f439ea8dfb9470a7467b7 : FAILURE in 1h 11m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b509c9c632304b2cbd7ea60a6cee3b7b : FAILURE in 1h 23m 36s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a07b25c238b49c28d4ec66b6e5ed2a4 : FAILURE in 1h 24m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1c3d820f85ee48abb515dbf5e73bf8ad : FAILURE in 1h 20m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a75c3042147f4c8abf54edf4ebfeea43 : FAILURE in 1h 09m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7a30c3916eb4ac3922ede201371330f : FAILURE in 1h 35m 59s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1923aaddc3a4e4ca8ea6cc6036f751e : FAILURE in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60e49d8d5fdc45fd8bd9bdf36ca92fb7 : FAILURE in 1h 14m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/743bed8fc72c40d18c9a647a32995e4c : SUCCESS in 2h 04m 42s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/827295e0faee476a9143625d006d6c7b : SUCCESS in 2h 45m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df964091f05b4c25b0b8b696b58e67ce : SUCCESS in 1h 56m 42s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2de7e618a82b47c1b0b85ab52646dec5 : SUCCESS in 2h 39m 27s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f7f193668a9c44ec830c68629af0fe4a : FAILURE in 2h 13m 03s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cf9219192be4e91a38568520e748247 : FAILURE in 2h 02m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759a71ff2cbb428e9578624fba318029 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"63f965f989cb4d104353e78f4cdbe5ee844134a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-13 13:25:20.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0f38275901b8b6ecb1981f38e3f0e19addf708ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:18:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4fd33a276cc46ae8dc1e73d3b151877\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d93266f9822247e8a6c4a8506756f1a1 : SUCCESS in 12m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/48f572d94747452e83cbbe2ef2d3aa62 : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/740ccf001ff34e00a72d1e35531c9a13 : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4a5040088eb40198ba5c815d30793ee : SUCCESS in 1h 34m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9ccf84208b304d03bcdf9d3d0bf5caa1 : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b29541687a4899a230f026fb4ef6d1 : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85640db0ce1b47da8e807b43376f70d1 : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecc3893e489c476c93902113c30eacac : SUCCESS in 1h 21m 05s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee373a015b4c4216ada59d100deaff7f : SUCCESS in 1h 19m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d73b56c906c4ab8b167ae11d3b98581 : SUCCESS in 2h 22m 29s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/810c43b8cf584a6e85cd48ae1108ad19 : SUCCESS in 2h 32m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ff2783e207b4779853ad4caa7849056 : SUCCESS in 1h 46m 19s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dab31a046b740b78e31f3bc0f9efdbf : SUCCESS in 1h 54m 22s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c791d9b117c74f4fb305f7827a3eb99c : SUCCESS in 2h 00m 52s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da14bf2acc48477f9f1276cd8ef76184 : SUCCESS in 2h 30m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bcb70d1af7a4aef9e831d5d0d5b01e8 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"f9c8cc7c8890bf319a34889c0e37cb31967744d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-14 10:25:58.000000000","message":"Topic osa/messaging_improvements removed","accounts_in_message":[],"_revision_number":4},{"id":"a69ecb166195860571a5b1926f377ee9a80e2c57","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-15 21:57:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3fb8ab13aba2bc2540bb4a884f9f58dc21d56902","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2024-05-16 04:09:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e038be5b49891d5cf54bcf5cc392edff04608bff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:11:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"630f7a963dc9942ba49a77221f332097a327837c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 06:03:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501c32d3de19408f9167e1bce671e15c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/748c00c714ec428ebb84f6122093751d : SUCCESS in 18m 39s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4235b59d5ab84495bfd9f2f4ddd63a9a : SUCCESS in 1h 23m 50s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/572646fae3854a059523770da2fc66ac : SUCCESS in 1h 32m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a3e861c15f63478d8794879489ae69c2 : SUCCESS in 1h 29m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1434a6d1ee794e309fe5709df283d5ad : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df17703bdf1e46cb9aa0e1b2abcf8a99 : SUCCESS in 1h 19m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10bc1ae00071450490112571533c028c : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1319a0ec065043b99b6622ecf4d8a859 : SUCCESS in 1h 08m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5e5486a0d34bff8cd5fca521787086 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":4},{"id":"047da180e775f98701738b8c275cbe7bab0c829a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 06:03:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"113826f6e25060091300f7e0b802ba75be13d820","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 06:04:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/096397e6e9014918aacb151700f1d68b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505fd31138a2491fa8e33fc59a3a4bae : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d9db42e5a95a6f7c0dbd6007e6e8a4447cf54e01","revisions":{"6777adf63a6328569cfc680c9c5694bf8d5cba41":{"kind":"REWORK","_number":1,"created":"2024-05-03 16:47:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/918103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/03/918103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/1"}}},"commit":{"parents":[{"commit":"7aa56eeb75211e925d9990406c2d593b6eef2204","subject":"Implement variables to address oslo.messaging improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7aa56eeb75211e925d9990406c2d593b6eef2204"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 16:47:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 16:47:32.000000000","tz":120},"subject":"Manage trove images through openstack_resources role","message":"Manage trove images through openstack_resources role\n\nThis unifies approach for common resource management, like image\nuploads.\n\nWith that we also split network creation with information fetch about it\nto include role only once and save execution time.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: Ib0126a0ac70aa613296a8d6e1ca61b34e22b02c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6777adf63a6328569cfc680c9c5694bf8d5cba41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/6777adf63a6328569cfc680c9c5694bf8d5cba41"}]},"branch":"refs/heads/master"},"1a4f95a7b95aca47e24ba9c33e975577d3664853":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-11 09:06:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/918103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/03/918103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/2"}}},"commit":{"parents":[{"commit":"5cf6094752d74f5af117339b77326e78b377b2bd","subject":"Implement variables to address oslo.messaging improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/5cf6094752d74f5af117339b77326e78b377b2bd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 16:47:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-11 09:06:04.000000000","tz":0},"subject":"Manage trove images through openstack_resources role","message":"Manage trove images through openstack_resources role\n\nThis unifies approach for common resource management, like image\nuploads.\n\nWith that we also split network creation with information fetch about it\nto include role only once and save execution time.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: Ib0126a0ac70aa613296a8d6e1ca61b34e22b02c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1a4f95a7b95aca47e24ba9c33e975577d3664853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1a4f95a7b95aca47e24ba9c33e975577d3664853"}]},"branch":"refs/heads/master","description":"Rebase"},"d9040f64c6b26d060c7b708d3bd189bc4b6818d1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-13 09:45:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/918103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/03/918103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/3"}}},"commit":{"parents":[{"commit":"41868739632a59f0bd0410e69744b21e3916efda","subject":"Implement variables to address oslo.messaging improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/41868739632a59f0bd0410e69744b21e3916efda"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 16:47:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-13 09:45:36.000000000","tz":0},"subject":"Manage trove images through openstack_resources role","message":"Manage trove images through openstack_resources role\n\nThis unifies approach for common resource management, like image\nuploads.\n\nWith that we also split network creation with information fetch about it\nto include role only once and save execution time.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: Ib0126a0ac70aa613296a8d6e1ca61b34e22b02c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d9040f64c6b26d060c7b708d3bd189bc4b6818d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d9040f64c6b26d060c7b708d3bd189bc4b6818d1"}]},"branch":"refs/heads/master","description":"Rebase"},"d9db42e5a95a6f7c0dbd6007e6e8a4447cf54e01":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-05-13 13:25:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/918103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/03/918103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/03/918103/4"}}},"commit":{"parents":[{"commit":"41868739632a59f0bd0410e69744b21e3916efda","subject":"Implement variables to address oslo.messaging improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/41868739632a59f0bd0410e69744b21e3916efda"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-03 16:47:32.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-13 13:25:20.000000000","tz":0},"subject":"Manage trove images through openstack_resources role","message":"Manage trove images through openstack_resources role\n\nThis unifies approach for common resource management, like image\nuploads.\n\nWith that we also split network creation with information fetch about it\nto include role only once and save execution time.\n\nChange-Id: Ib0126a0ac70aa613296a8d6e1ca61b34e22b02c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d9db42e5a95a6f7c0dbd6007e6e8a4447cf54e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/d9db42e5a95a6f7c0dbd6007e6e8a4447cf54e01"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
