)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~920455","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~If92f81bee0ffd3556e1b1e1f26e7572af04ff1b2","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/messaging_improvements","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-05-31 15:45:08.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-05-31 13:51:04.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"If92f81bee0ffd3556e1b1e1f26e7572af04ff1b2","subject":"Define unique hostname for QManager","status":"MERGED","created":"2024-05-25 13:38:12.000000000","updated":"2024-05-31 15:46:03.000000000","submitted":"2024-05-31 15:45:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"920455-osa/messaging_improvements","meta_rev_id":"a3503b367b5c3e348f73529f2d835392ff7b8bbb","_number":920455,"virtual_id_number":920455,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-31 15:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-05-31 12:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-31 13:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-31 13:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-25 16:37:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-26 20:47:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-05-31 12:42:44.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-05-31 13:51:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"06dcae276a0037ac8030de733e33b9e9fe69c496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-25 13:38:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"460313a3ed82545fca29207ed81651f0bcfb5082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 16:37:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf547cb1fa5466992a1aae26d7b4cdf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3728454d20b40d69b2fa693a18d28e7 : SUCCESS in 11m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4cb6c09f69db422ba007df655cc052ee : SUCCESS in 1h 22m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/57080a878f414b9f8c66511a073cf18b : SUCCESS in 1h 23m 51s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/62a701e360c548dab34c44468c8fc7f1 : SUCCESS in 1h 13m 01s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ff172db465cf42c9b97353e242a1afc9 : SUCCESS in 1h 31m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7cbbf3d6b8834635b2eb1e1a4d10a7df : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e6eb56acda2e465dabec665fac36522d : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2dde1afdb6e49c994ec84ec00fb5551 : SUCCESS in 1h 11m 28s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/399e4164e4734b18b4352fd3a1c4a31b : SUCCESS in 1h 29m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fe41d80cf32a4e699f50b1594444c0d0 : SUCCESS in 2h 08m 36s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/206c73b71cbd4c1089ea97d1bd124e9d : SUCCESS in 2h 35m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ff04fcf204b4aa9b753a701c57586b7 : SUCCESS in 1h 49m 55s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f44ee7903cb541579d752120205e89b3 : SUCCESS in 2h 21m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d20a71ca9b474853a47301e27853afa1 : SUCCESS in 2h 07m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef0ce40511ce4990b6ff17ce0a9210ef : SUCCESS in 1h 38m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed11931c40043819648002b57eeb650 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9cecb4e061dc7d09f2ad55e2fad36a270255a3fc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-26 20:47:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de107714457856581783691cee655c6eaa9596c7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-29 07:06:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14ca4896cf7e507531462978137727939bb128ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-29 07:50:21.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2796d2e0bb57782e9e7c3ff44c861c806f8f6c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 10:52:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c90ff8301c645069d076ed0f7194096\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bc3a39fbf7e4d6cb4a781844fa912fc : SUCCESS in 14m 13s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fac7737a54e7441fb2ad0c70d5b47d8c : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b74963512f954150b4bb971f756be7bd : SUCCESS in 1h 05m 11s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b36915bd9bc4e1e91cabbf0da6d1e66 : SUCCESS in 1h 26m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9d30a6c8b33d48d389e234b7338d9ddc : SUCCESS in 1h 42m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d404bb39df904b818ca7ed41eed7602f : FAILURE in 32m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/12c39f12305246e5bd59bebd6137ea02 : SUCCESS in 1h 38m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/214bf64f9c4f4d3faf268e9ad83db2f8 : FAILURE in 31m 11s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c2d45cd35594e6db0da539858690183 : FAILURE in 28m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/df577cc8125644ca8a79744425f5a57e : SUCCESS in 2h 20m 07s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51d518f4933e40f4b28ef57ba4c537e0 : SUCCESS in 2h 44m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0389589d64424e6c882b09e745c66d20 : SUCCESS in 2h 14m 19s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d654604bd414d018904984121f818c9 : SUCCESS in 2h 07m 31s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cabc5acc18b141c2b208acbc9d012084 : SUCCESS in 2h 23m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91300ad58dc9444fa00c69494127c394 : FAILURE in 40m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75b7c1626a4446d1bd45b0c8652aa324 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"db5402abecfb12d0944137b0e271079015a9422d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-30 16:46:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4bba0422487cbe2bb2439b0cece1c1000bdf386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 19:26:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a72542bd56f442c99888b87b18ddde8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3ddcfaa41304263a7be3ea4e6235557 : SUCCESS in 12m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/dfba4298bd414463828bab52ba6ac9be : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/da41169808c446f78d5f3fede7fae250 : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/af3148cc6c6a457490512a439351a7cf : SUCCESS in 1h 28m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ee667624ea164175bc641cd4dc8444fe : SUCCESS in 1h 23m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef0af627241d4bbda11f814e86df8761 : SUCCESS in 1h 14m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84315db76bbd411daeee64a267a06e89 : SUCCESS in 1h 43m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8c68396f1d74002bddd6a056851c32b : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/268e3563cafa40cc8de31cc15bd00217 : SUCCESS in 1h 17m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b8e879206e434711b24658761d50b89c : SUCCESS in 2h 15m 43s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a8b2119e26584049b939dee3d7daa7d3 : FAILURE in 2h 21m 55s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/143ecc366fdb4b9f8d57816a4646b3d3 : SUCCESS in 1h 39m 28s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d7d32b557ce46ba999a1f0afeb6e483 : SUCCESS in 2h 20m 39s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0452984cff134befb68c5e404e84be71 : SUCCESS in 2h 09m 49s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa0dcf48d9fc4b6eaa47fa6d033a923f : SUCCESS in 1h 45m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f432119b916a43f7bad3e4751cd5fa11 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"dcd1a2cd9180cdc2480fa430360914937a2891fe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-31 04:54:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7532abf5c66988c5d1e77c63c5e19205c38984e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 07:38:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33dfc7e06cf74e6babde99d8b5721f79\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d5db16ab19249fbb79e408868f72317 : SUCCESS in 10m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9503147dc31c4cbcbf8527619ecada0e : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/545cf3d72f84454a8df0a1fcc6d14b43 : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e369b4ae69de4f7bb9d8851d263c0e15 : SUCCESS in 1h 14m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e345eb939d8e47059df124a14e98e226 : SUCCESS in 1h 32m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd4ef2597a6f4fe795304a41c47ccfac : SUCCESS in 1h 14m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf081e72b0d6488ab5ce9172784ec0a2 : SUCCESS in 1h 25m 25s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a931c7066564a24b5b4489801ebb08c : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/840ddb9ef7e943808163bd242e5d3acf : SUCCESS in 1h 22m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b09bff2f417242b08d99ad3e28216c2d : SUCCESS in 2h 07m 06s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19bb758f10ba498ab0952da4f9e67769 : SUCCESS in 2h 07m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/483d15613f7e440d8d241943104eb090 : SUCCESS in 2h 11m 15s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3be8eada01994180ad5f3d29dd36ce6f : SUCCESS in 2h 37m 32s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd5f649412e34fc098f553e1ba569cb3 : SUCCESS in 1h 48m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c7e91e196ca4b539fc09e2b187aa328 : SUCCESS in 1h 59m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5e21d6ddc24ee795a4befeaad5d0fd : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d536ef74c14e9fbd9bb9d4ed279d3bc0bd3e15cc","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-31 12:42:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a89a88eed411951c61a85b2899a36454f2b145da","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-31 13:51:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2453a3837e9adbd6282a152691e93d1b925a586c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 13:53:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f49ce5e6ee963e85e79937f9903375b9d63cc5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 15:45:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/134ee4f494ea448c8f814aa30b4fe0ed\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d113cd6de9941a0964f12b9d94434c3 : SUCCESS in 11m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ff1e84268eb48e68113bdcfb0bcb454 : SUCCESS in 1h 20m 18s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2bcdc78466a84a8fa4ee922e429a8a37 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/118efb28b21c4376934cbbc143fa2ccf : SUCCESS in 1h 20m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9857bdde1f8d4a12b31bf4b3da9dd49f : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d47d784ff94c5cb6bb3949db72a4e6 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b6d1a5e712464749b254dc293e117c5a : SUCCESS in 1h 37m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efe870581e454897bd8e40c361093c5b : SUCCESS in 1h 09m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae6d1dd4907b4a1c9192c0bd5c35a95f : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8bc707bf73c4ed6eb49af61133c18c0bb2f2cf22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 15:45:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a3503b367b5c3e348f73529f2d835392ff7b8bbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 15:46:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22b9a056b574d16b183d298a1d9bed7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8781e97fb30453fbb829e780740c67e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"428a647bdc5341a3d1c5bd5e22bfc4dc2d96a10f","revisions":{"238aeb7e359a7b100c6187a191573040d27f8706":{"kind":"REWORK","_number":1,"created":"2024-05-25 13:38:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/920455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/55/920455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/1"}}},"commit":{"parents":[{"commit":"37097eacb3a0da2e9a8fe08474998a6e99c5ef29","subject":"Merge \"Add tag to enable targeting of post-install config elements only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/37097eacb3a0da2e9a8fe08474998a6e99c5ef29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-25 13:38:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-25 13:38:03.000000000","tz":120},"subject":"Define unique hostname for QManager","message":"Define unique hostname for QManager\n\nDue to the shortcoming of QManager implementation [1], in case of uWSGI\nusage on metal hosts, the flow ends up with having the same\nhostname/processname set, making services to fight over same file\nunder SHM.\n\nIn order to avoid this, we prepend the hostname with a service_name.\nWe can not change processname instead, since it will lead to the fight\nbetween different processes of the same service.\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/2065922\n\nChange-Id: If92f81bee0ffd3556e1b1e1f26e7572af04ff1b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/238aeb7e359a7b100c6187a191573040d27f8706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/238aeb7e359a7b100c6187a191573040d27f8706"}]},"branch":"refs/heads/master"},"428a647bdc5341a3d1c5bd5e22bfc4dc2d96a10f":{"kind":"REWORK","_number":2,"created":"2024-05-29 07:50:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/55/920455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/55/920455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/55/920455/2"}}},"commit":{"parents":[{"commit":"37097eacb3a0da2e9a8fe08474998a6e99c5ef29","subject":"Merge \"Add tag to enable targeting of post-install config elements only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/37097eacb3a0da2e9a8fe08474998a6e99c5ef29"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-25 13:38:03.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-29 07:50:21.000000000","tz":0},"subject":"Define unique hostname for QManager","message":"Define unique hostname for QManager\n\nDue to the shortcoming of QManager implementation [1], in case of uWSGI\nusage on metal hosts, the flow ends up with having the same\nhostname/processname set, making services to fight over same file\nunder SHM.\n\nIn order to avoid this, we prepend the hostname with a service_name.\nWe can not change processname instead, since it will lead to the fight\nbetween different processes of the same service.\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/2065922\n\nChange-Id: If92f81bee0ffd3556e1b1e1f26e7572af04ff1b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/428a647bdc5341a3d1c5bd5e22bfc4dc2d96a10f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/428a647bdc5341a3d1c5bd5e22bfc4dc2d96a10f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
