)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~924642","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Iacbddd8bc0c5f7d38ca85a3ef4df536ea4f66782","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/frist_host_refactoring","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-10-07 12:24:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iacbddd8bc0c5f7d38ca85a3ef4df536ea4f66782","subject":"Ensure that first/last host detection is deterministic","status":"MERGED","created":"2024-07-22 14:25:20.000000000","updated":"2024-10-07 12:25:53.000000000","submitted":"2024-10-07 12:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924642-osa/frist_host_refactoring","meta_rev_id":"c5d5c32ccb1a6d4d1594ce25f3dd0d51f7e6f518","_number":924642,"virtual_id_number":924642,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-07 12:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-10-04 09:17:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-07 10:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-07 10:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-22 16:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 09:17:32.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2024-10-07 10:01:40.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"cf3c019857d31b0ff048a39862abd81bf8626ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-07-22 14:25:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"429de6bd0ffdd6f9d167e8db69c48994d4d2354a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:32:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/700ef0161aab469d8cd649e0609daa71\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2fc0a2b8bfc456898a3e2e01d1286fc : SUCCESS in 11m 06s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/73f0e036603f4ced89dbcc36d864037c : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2de2f7861fb34db8a7b9898eefd18f78 : SUCCESS in 1h 20m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8338414a14a9426dbb030787308de30d : SUCCESS in 1h 37m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/260be54f86b340ebba356d54918e9ba1 : SUCCESS in 1h 19m 09s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f454f57f26524ceeb3df5587857e3137 : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b12560925dab4889a148a4b4456ef2ea : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f578c496736741f595c6c167a95a1187 : SUCCESS in 1h 13m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dafe21151d1f43d0b448a3e9c6824db8 : FAILURE in 2h 03m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87d3f202f8ef434a8a1811aa523edf9f : FAILURE in 1h 48m 23s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9d8f04afad8467184b271a206c0df69 : FAILURE in 1h 56m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d658bf1a8d2c417e9a1fff41088f6006 : FAILURE in 1h 40m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469730ef83f5499a92c7198664b322b8 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d0ee68d4161ef43a6973775205687d58228222e1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-03 15:16:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7be7ccc431e8d2256e078fdef7e6aa9046aad725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:36:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7b431b0b30d436fb9b2f5007c4f8120\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7031a5f484db4a0ba3cd948459a30760 : SUCCESS in 9m 21s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/43bc31a2becb48999e4f27de7c90c373 : FAILURE in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf49b9455a0470f8947292a392aaa57 : SUCCESS in 1h 23m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aecdb6285e9f439091aa1f479518ecec : SUCCESS in 1h 19m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7b5d3d670a34cc38ef6c122d40bc257 : SUCCESS in 1h 16m 29s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3a499b43e7df4324819378201ae781df : SUCCESS in 1h 26m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c955774087a54083a7df4e366cd398aa : SUCCESS in 1h 20m 52s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f91e8837f31434a92e58e17d3403143 : SUCCESS in 1h 07m 44s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c03c75835ba24f90a57b969b2b674ed3 : FAILURE in 1h 56m 20s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25b401b693634b63a2cda25bd77fba96 : FAILURE in 1h 40m 02s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b49ff87508eb48ca81526e6e820683d2 : FAILURE in 19m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef0437e4698744768655647fd1e03c1f : FAILURE in 1h 35m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210107b08f034ceb87d585aa8601324f : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"25137781ae6f17060995377c2d99ba2213838695","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-03 11:38:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e4052fd3367ccb42c4ce20b24613efd6c23772c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 14:49:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b0cacd210e44460b3c6c8f0799980be\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10af3cfed17e402ea0d5f3282e5fb898 : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/1ee0ba379edd4d689ec8bfe1adcd516e : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e9c10120f74423581c75e94d7cf88dd : SUCCESS in 1h 31m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8f1fb2e918784bccabc0ab67f259b03f : SUCCESS in 1h 18m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3934068357e84d02bc0c00ef1646cc51 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13d492eab48f432797031fee9620b5c1 : SUCCESS in 1h 28m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b9c9069b2be14fbaab77a905bdf08ea6 : SUCCESS in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eefe4727bde49378caaa703e44a3f3f : SUCCESS in 1h 25m 55s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0560b5d9d2ca4ed6ba571d3a5d7ea51e : SUCCESS in 1h 38m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/658df8095eac49339114cbcb42635896 : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31ceec9af5f14ddaadebf347229e3cd9 : SUCCESS in 1h 26m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5cdd162b56884fd9883251085eeded7f : FAILURE in 1h 50m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/921357f2be894f3c99cb3a7844bb3489 : TIMED_OUT in 3h 03m 13s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c1bb93e68f2340ab9abade8eb0d1223b : FAILURE in 1h 59m 43s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c588e522c9045d5993a8cb9a72d3504 : FAILURE in 2h 20m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1fe9a5348f460db3e2cc171ab59226 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"73d848bcf67e26b3e1b65a718e3d55f9f32301ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-03 15:29:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0df6fe13e644fde96258d55b82594d8a965fff13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 17:45:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2918d902f9e8467eb9126bceb0b59799\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e48cfd3e4124c0c999be476d741feae : SUCCESS in 6m 47s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b39a7b1e2da42daa7253f40e6677f62 : SUCCESS in 2h 12m 32s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f9e33338d984121bc94fc6b69821bce : SUCCESS in 2h 04m 52s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/557fb279ec5d416aa3f83b526a9a73ab : SUCCESS in 2h 08m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5589724fa2784b0b89eb501bb23bfaef : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6a1fd7d04da485686f9566b0cfcd74a : SUCCESS in 1h 15m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a4011eab62142c497e276afe4716cb7 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bac653850559df8aa70a4f3598cfcb55b02959cb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-10-04 09:17:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"692506db69c2abd469db5f90ef5bea128ea1996e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-10-07 10:01:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e624ebf28a65e26fb41d14d3e51c2076f2a606e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 10:04:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fdf34932f3e75c22b00a1fce37d286f3e38f7285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:24:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25f73edaf44d415b88a7a42e20fb6e16\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dd41a5575fd4d26a257721bdbb7a899 : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/80a4da9c1f0244cc8033be0b8d49c9f1 : SUCCESS in 1h 55m 46s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9c88571aa1f4a73a26ce8f3d21ae6ee : SUCCESS in 1h 53m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afb2fa6922aa45caa3a916513db081d3 : SUCCESS in 1h 45m 20s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/064856740f0c4576a689486fbd6783e7 : SUCCESS in 2h 15m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb6d5e97359445182c3a3538be94656 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0ff9a968f89f522f6528de3bcc13deed2a2fc2e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:24:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c5d5c32ccb1a6d4d1594ce25f3dd0d51f7e6f518","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 12:25:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/966b5fbb938148b7a7f5ba97f6f18951\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2dac8957b0847b299a4cfa0147d28ad : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a73e4a0b854a07b308d17a795e24afd79d0dff8","revisions":{"09551e82a0204d7cabc34b28cc5d2b326dd600f0":{"kind":"REWORK","_number":1,"created":"2024-07-22 14:25:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/924642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/42/924642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/1"}}},"commit":{"parents":[{"commit":"4d141d89034165792c4fd37a325458770fb57a5b","subject":"Merge \"reno: Update master for unmaintained/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/4d141d89034165792c4fd37a325458770fb57a5b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-07-22 14:25:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-07-22 14:25:13.000000000","tz":120},"subject":"Ensure that first/last host detection is deterministic","message":"Ensure that first/last host detection is deterministic\n\nWith ansible-core 2.16 a breaking changes landed [1] to some filters\nmaking their result returned in arbitrary order. With that, we were\nrelying on them to always return exactly same ordered lists.\n\nWith that we need to ensure that we still have determenistic behaviour\nwhere this is important.\n\n[1] https://github.com/ansible/ansible/issues/82554\n\nChange-Id: Iacbddd8bc0c5f7d38ca85a3ef4df536ea4f66782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/09551e82a0204d7cabc34b28cc5d2b326dd600f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/09551e82a0204d7cabc34b28cc5d2b326dd600f0"}]},"branch":"refs/heads/master"},"9a73e4a0b854a07b308d17a795e24afd79d0dff8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-03 15:29:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/924642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/42/924642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/42/924642/2"}}},"commit":{"parents":[{"commit":"5b668b373c6a6909327cce08dbfcf26a9bb508d7","subject":"Disable upgrade jobs for Trove","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/5b668b373c6a6909327cce08dbfcf26a9bb508d7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-07-22 14:25:13.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-03 15:29:29.000000000","tz":0},"subject":"Ensure that first/last host detection is deterministic","message":"Ensure that first/last host detection is deterministic\n\nWith ansible-core 2.16 a breaking changes landed [1] to some filters\nmaking their result returned in arbitrary order. With that, we were\nrelying on them to always return exactly same ordered lists.\n\nWith that we need to ensure that we still have determenistic behaviour\nwhere this is important.\n\n[1] https://github.com/ansible/ansible/issues/82554\n\nChange-Id: Iacbddd8bc0c5f7d38ca85a3ef4df536ea4f66782\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9a73e4a0b854a07b308d17a795e24afd79d0dff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9a73e4a0b854a07b308d17a795e24afd79d0dff8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
