)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~938461","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I4171afffa496aca00e9892b596ac85d2e978665b","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/retire_qdrouterd","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-01-08 16:32:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4171afffa496aca00e9892b596ac85d2e978665b","subject":"Remove support for amqp1","status":"MERGED","created":"2025-01-06 10:54:17.000000000","updated":"2025-01-08 16:33:17.000000000","submitted":"2025-01-08 16:32:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938461-osa/retire_qdrouterd","meta_rev_id":"5b20a3939c270b4460c798d74c1b2af49f718556","_number":938461,"virtual_id_number":938461,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-08 16:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-01-08 14:18:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2025-01-08 10:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-01-08 14:18:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 13:25:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-08 10:26:46.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-01-08 14:18:54.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"53146faed3675fd2ea533722854905593a58979c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-06 10:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd34d2bc9a53af527d028edd8c17e836c47628f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 13:25:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/395246936f4e44deb6ef661ffba1720a\n\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0639456a39974d268e9fa68442cf4d5b : SUCCESS in 2h 12m 06s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3d2999155fa74cdda7a7d63baf2dc66c : SUCCESS in 2h 17m 14s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7b5b79ac8b8143ea9db63d09128901ad : SUCCESS in 2h 07m 01s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d30cc60c2e02423c815bfd28bcc73b71 : SUCCESS in 1h 45m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4ec6481fb4e4cb7a83b0fe8f94b539b : SUCCESS in 1h 52m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc23bf2020c4a98b7828ec55bd520d0 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bc223ed22c91795768cc8085b5e3e82906d5ed42","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-08 10:26:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c26517f0b5f4bdad8ab65616aea818fa008a6f5","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-01-08 14:18:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9991324f91e6b6862807d403face521e18fa903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 14:21:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37c37dfef5ea2c1c800e265e7ae23b34ed7b6639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 16:32:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a83923d7c4a64d2c94c86aae0ad8d451\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/183769bfb82f4f1c9247bb766b856b4f : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd32577bfe234934a8127840f964e432 : SUCCESS in 2h 03m 47s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/12483a4babf84649b1ffd59f15c10c95 : SUCCESS in 1h 53m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/601db99cf0e347fe9c490600311fbd3c : SUCCESS in 1h 51m 59s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82f641df3f3842a8900d70904e07352a : SUCCESS in 1h 49m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11e17cda7474328ab1bc133717e3820 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c20aab42eb2cedf027d7b383bdf40689da74ed51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 16:32:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b20a3939c270b4460c798d74c1b2af49f718556","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 16:33:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e8f545237a459c923b99c586831c0b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592fb9db36c1482a9239baf1f7972278 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7efed51b5a13a546c5311dd8e88bcdd30178a3bc","revisions":{"7efed51b5a13a546c5311dd8e88bcdd30178a3bc":{"kind":"REWORK","_number":1,"created":"2025-01-06 10:54:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/61/938461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/61/938461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/61/938461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/61/938461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/61/938461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/61/938461/1"}}},"commit":{"parents":[{"commit":"9a73e4a0b854a07b308d17a795e24afd79d0dff8","subject":"Ensure that first/last host detection is deterministic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/9a73e4a0b854a07b308d17a795e24afd79d0dff8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-01-06 10:54:04.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-01-06 10:54:15.000000000","tz":0},"subject":"Remove support for amqp1","message":"Remove support for amqp1\n\nSupport is removed in oslo.messaging so we remove support in\nopenstack-ansible roles.\n\nChange-Id: I4171afffa496aca00e9892b596ac85d2e978665b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7efed51b5a13a546c5311dd8e88bcdd30178a3bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7efed51b5a13a546c5311dd8e88bcdd30178a3bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
