)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~941433","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~If8709a7ca55857c6a97deed6c19b5fc3503688ce","project":"openstack/openstack-ansible-os_trove","branch":"master","topic":"osa/lint_autofix","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-02-19 21:55:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If8709a7ca55857c6a97deed6c19b5fc3503688ce","subject":"Auto-fix yaml rules","status":"MERGED","created":"2025-02-12 14:17:02.000000000","updated":"2025-02-19 21:56:31.000000000","submitted":"2025-02-19 21:55:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941433-osa/lint_autofix","meta_rev_id":"86a2c640052b82c7af5eaadbc7a64704bccbf4d2","_number":941433,"virtual_id_number":941433,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 21:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2025-02-19 08:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-17 17:16:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2025-02-19 08:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 17:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-17 17:16:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-02-19 08:28:45.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"ad42649f7f0fbda6887816d8c9df6f2daa7fe7a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-02-12 14:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d80d393cd03d54bf68c28f8f377ddf2c2bb9819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 17:00:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5592d44840ae437c8414a33412700194\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6681c92483d6489599a65ffc8b4143bf : SUCCESS in 9m 21s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a93b07f99d9347698c51be355cd4b947 : SUCCESS in 2h 10m 05s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e941f52cef1a43fea512679b1added2e : SUCCESS in 2h 20m 49s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/9beb8f75c25a4ac09d738f3c5559362f : SUCCESS in 1h 56m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89347b40391145229aa9975c3e779658 : SUCCESS in 2h 04m 06s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9fde5709fc6462793dbb741be2c2c8b : SUCCESS in 2h 30m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804b42e7ce01401d8864f40337ffeb1c : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"948cfc9b7d749b1318a15eb30ed256f84b29fd7d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-17 17:16:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc07d09eb18821eda8b5e26803ac0cb914b5cdcd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2025-02-19 08:28:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c31be9c965e91eace71fca385cb619302434ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 08:30:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f7fd03fd2a9aaa08fccbbdd535ca41de09006af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 10:40:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94ed8803f23d4f078d5540d4c25cb18c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d96d309c0bd42ceaad2d0fa4bb5cb01 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/39510eea4ab2452aba4b1254f0471362 : FAILURE in 1h 56m 14s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3e530835b2834c0ba313ea3a1740f258 : FAILURE in 1h 54m 48s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d60eea4e02f47208a9317e724f60ca0 : SUCCESS in 1h 47m 44s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fabd2601d3d4918b241aa5dd2dc21c8 : FAILURE in 2h 05m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c6ce399ef1448f9564e43fadf88708 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"58afbc5fd41c0d71743e09b2aea53becd20db3ba","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-02-19 17:25:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"930003d5796344a2ffc69e0a0c75e0039948c686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 19:45:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cfcf7fe2d70450bba8c23113917f99b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13362a88f85949ddb7a1d4727d841f2f : SUCCESS in 7m 16s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec53684653c74c708d7511716ddb7c4f : SUCCESS in 2h 00m 47s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c1a62150b9944e6682971f433765c516 : SUCCESS in 2h 04m 15s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cfccc21417ae4c1789fb314e665f2af3 : SUCCESS in 1h 58m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/723ef14a47f84d80ac74d0d8385da7ed : SUCCESS in 2h 02m 20s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/565f0292612a49e98b1da91312226c46 : SUCCESS in 1h 55m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68ff5976e634aada981b2be82bd7f85 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8d1ffde18b39c114eb9292f78027ffebcedf8562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 19:47:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75549a87e4ebea307e322ec8d67fdc186e31849f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 21:55:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6054f4d5adf4b839a9fafac9750f002\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa02404a87404419b38796dc4afff854 : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c8f7344dea11406f88c944416357ee96 : SUCCESS in 2h 05m 58s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/961e241e0708470ba1b82cf20cde36bc : SUCCESS in 2h 01m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca7539266bb84b0cba6873d4d4e8789f : SUCCESS in 2h 07m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9e8211be20f4124bed70944539f062c : SUCCESS in 1h 55m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0bfaf98b5b34a15a8d3039d0d87d23f : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6f3d659469e694787230647fac4fd4b96ed7232f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 21:55:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86a2c640052b82c7af5eaadbc7a64704bccbf4d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 21:56:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aa986ad47994f45bb2ad59ab7981088\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3256a822b51d49cc9dfa263d774078f2 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8007941f22c67462baef0386c1611581accf3897","revisions":{"8007941f22c67462baef0386c1611581accf3897":{"kind":"REWORK","_number":1,"created":"2025-02-12 14:17:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/33/941433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/33/941433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/33/941433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/33/941433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/33/941433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/33/941433/1"}}},"commit":{"parents":[{"commit":"7efed51b5a13a546c5311dd8e88bcdd30178a3bc","subject":"Remove support for amqp1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7efed51b5a13a546c5311dd8e88bcdd30178a3bc"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 14:16:53.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-02-12 14:16:53.000000000","tz":60},"subject":"Auto-fix yaml rules","message":"Auto-fix yaml rules\n\nIn order to reduce divergance with ansible-lint rules, we apply\nauto-fixing of violations.\n\nIn current patch we replace all kind of truthy variables with\n`true` or `false` values to align with recommendations along with\nalignment of used quotes.\n\nChange-Id: If8709a7ca55857c6a97deed6c19b5fc3503688ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/8007941f22c67462baef0386c1611581accf3897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/8007941f22c67462baef0386c1611581accf3897"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
