)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~951590","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I29984c565c808bca79ef39a02e79d9b015a70786","project":"openstack/openstack-ansible-os_trove","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-09 18:21:07.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-06-09 15:54:41.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I29984c565c808bca79ef39a02e79d9b015a70786","subject":"Add support for generation of SSH keypairs","status":"MERGED","created":"2025-06-02 19:06:55.000000000","updated":"2025-06-09 18:22:08.000000000","submitted":"2025-06-09 18:21:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951590","meta_rev_id":"ef1d791210056e8b8595a6bfd254ab79fa058169","_number":951590,"virtual_id_number":951590,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 18:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"all":[{"value":2,"date":"2025-06-09 15:18:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-09 15:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-09 15:54:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-02 21:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-09 15:18:09.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2025-06-09 15:36:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-06-09 15:36:42.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"3fcbf47169f82aedea71e0880421d72281fab6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 19:06:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4dfa8e31a10717749f12fb61d0f9d0bfbf640a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 19:07:33.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fc44c41033020d1c75ef2fec70ed4ab1caefee35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:41:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c12964f5c44446ad444cca23c3f92d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98ba87b4dde84df58e3afa66e9560fa6 : SUCCESS in 8m 39s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/178bc4412e63469a98ec59932720456a : FAILURE in 1h 54m 15s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdf3a0e3fe9d4f629dd91e3626f7edd9 : FAILURE in 1h 48m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dca60b02a64258b84db6800b8e7514 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/546185f4617c4a48b629137a5eed881e : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2},{"id":"acb01f8b29b34f3bd2758ac1438d20af7fbf66e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 10:51:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1fb7d32d41467df73fa125204e65669409e2fe95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:22:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67607793c6df409b91fe595c32a02a78\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f129accefb9941048ee80927fc086e4c : SUCCESS in 5m 41s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/704e0eb794404b2597a98783c490901b : FAILURE in 1h 54m 20s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39ca8da536264f1ba008f443e0bb4ccc : FAILURE in 1h 46m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4cd3b8732948bdbe3971c3f94e350b : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/083eabce012a4a85a7a1f168a65fe29d : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a4db9e4a7d989cd731e9ec2fc74343ba44bd75f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-03 19:56:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"da29f9c694f77883de791c68b00e972d216a41c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 22:44:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fb53209a3ff49a5b36337d8402d58ee\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70f9434d59844877a430640ca4843817 : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c9f8b4b76516437394c1449d69da4585 : FAILURE in 2h 11m 12s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ebde0e17aa64546a7eca5e75d0b0960 : FAILURE in 1h 40m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0383720b1002489dac617cdb2c2dd4d6 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2890850f333c4bc6a465ebae5119e792 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":4},{"id":"65a7b1969a6cf3362a45d3eb1075057824ce9859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-04 11:53:50.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5eb7f287ef3b34ca6b9ec49ccaf0db9066829b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 14:20:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/622b1e9d0eb640e086df20c465e1ecbe\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97b4a2d7cea34d328b3d46756926d630 : SUCCESS in 6m 29s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/cfee0ea143a74be293f38e3b378f1771 : SUCCESS in 1h 52m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/359939cd016e4e6fa36273443adcf174 : SUCCESS in 1h 55m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c9b90bfea04a7d8aeedc7c0e8d8918 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef296b363adc4d1e8f13901e72b34481 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":5},{"id":"34901dde17963f5c798d4022bcf025d2e37ed0fd","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2025-06-09 15:18:09.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ce80904929646f9c9e757ed00a8f8886867d279e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-09 15:54:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1be83a19c369f01f38304f231b9eeeefb3e3571a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:56:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fc5d9fdd3f943c80b74951cd8a509074f4d052f2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-09 16:14:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b8af428dce96b87f2404b04cb647c506217f2827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:21:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f092f9f2ca4a4f74b51692e76a6a8c9f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24290a1071524b74a598df4d2c0e9422 : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/20d0cfb1619d446d8e2b42aefe3a40f0 : SUCCESS in 2h 01m 17s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d5e00a897b44838bc7f11e676c5b188 : SUCCESS in 2h 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33066f8462d9427abfe2f778c269143e : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21c30ae5ec984ed68e73cabf5226a773 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":5},{"id":"f39af72edf7d83d8876660d22e8d62f23d30adc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:21:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ef1d791210056e8b8595a6bfd254ab79fa058169","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:22:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54ea0a05b6ac4ea981072f2cd68957d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee31e29d4c04a6d8687534f78e3e10c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3035e2533323e7ab42c75ed65f77ff2f739323c0","revisions":{"2dc4bcd05fd35995e43376abaf692eb923a4985c":{"kind":"REWORK","_number":1,"created":"2025-06-02 19:06:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/951590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/90/951590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/1"}}},"commit":{"parents":[{"commit":"46862920df250163ca624eca834dd9582752ca90","subject":"Do not use group as a \"setup_host\" variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/46862920df250163ca624eca834dd9582752ca90"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-02 19:06:17.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-06-02 19:06:22.000000000","tz":120},"subject":"Add support for generation of SSH keypairs","message":"Add support for generation of SSH keypairs\n\nIf operator needs to enter guest agent instance for troubleshooting,\nthey need to create a keypair, upload it as Trove user specifically,\ncreate a security group and define it\u0027s UUID in trove.conf\n\nThis process is quite cumbersome and time consuming, thus this\npatch implements mentioned fuctionality and can be enabled\nwith a single variable.\n\nProcess is alike to what we have and been using for Octavia for a while.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: I29984c565c808bca79ef39a02e79d9b015a70786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2dc4bcd05fd35995e43376abaf692eb923a4985c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/2dc4bcd05fd35995e43376abaf692eb923a4985c"}]},"branch":"refs/heads/master"},"61976412c7aaa179a33a19a676b9e99e17ae701f":{"kind":"NO_CHANGE","_number":2,"created":"2025-06-02 19:07:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/951590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/90/951590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/2"}}},"commit":{"parents":[{"commit":"46862920df250163ca624eca834dd9582752ca90","subject":"Do not use group as a \"setup_host\" variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/46862920df250163ca624eca834dd9582752ca90"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-02 19:07:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-02 19:07:21.000000000","tz":120},"subject":"Add support for generation of SSH keypairs","message":"Add support for generation of SSH keypairs\n\nIf operator needs to enter guest agent instance for troubleshooting,\nthey need to create a keypair, upload it as Trove user specifically,\ncreate a security group and define it\u0027s UUID in trove.conf\n\nThis process is quite cumbersome and time consuming, thus this\npatch implements mentioned fuctionality and can be enabled\nwith a single variable.\n\nProcess is alike to what we have and been using for Octavia for a while.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: I29984c565c808bca79ef39a02e79d9b015a70786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/61976412c7aaa179a33a19a676b9e99e17ae701f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/61976412c7aaa179a33a19a676b9e99e17ae701f"}]},"branch":"refs/heads/master"},"1c2fef6199089f2e40273cc13aac9c4981bf2ff2":{"kind":"REWORK","_number":3,"created":"2025-06-03 10:51:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/951590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/90/951590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/3"}}},"commit":{"parents":[{"commit":"46862920df250163ca624eca834dd9582752ca90","subject":"Do not use group as a \"setup_host\" variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/46862920df250163ca624eca834dd9582752ca90"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-02 19:07:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-03 10:16:36.000000000","tz":120},"subject":"Add support for generation of SSH keypairs","message":"Add support for generation of SSH keypairs\n\nIf operator needs to enter guest agent instance for troubleshooting,\nthey need to create a keypair, upload it as Trove user specifically,\ncreate a security group and define it\u0027s UUID in trove.conf\n\nThis process is quite cumbersome and time consuming, thus this\npatch implements mentioned fuctionality and can be enabled\nwith a single variable.\n\nProcess is alike to what we have and been using for Octavia for a while.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: I29984c565c808bca79ef39a02e79d9b015a70786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1c2fef6199089f2e40273cc13aac9c4981bf2ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/1c2fef6199089f2e40273cc13aac9c4981bf2ff2"}]},"branch":"refs/heads/master"},"ce76bece83063bef3f6637e6d550a46d5c92bd7d":{"kind":"REWORK","_number":4,"created":"2025-06-03 19:56:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/951590/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/90/951590/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/4"}}},"commit":{"parents":[{"commit":"46862920df250163ca624eca834dd9582752ca90","subject":"Do not use group as a \"setup_host\" variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/46862920df250163ca624eca834dd9582752ca90"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-02 19:07:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-03 19:55:55.000000000","tz":120},"subject":"Add support for generation of SSH keypairs","message":"Add support for generation of SSH keypairs\n\nIf operator needs to enter guest agent instance for troubleshooting,\nthey need to create a keypair, upload it as Trove user specifically,\ncreate a security group and define it\u0027s UUID in trove.conf\n\nThis process is quite cumbersome and time consuming, thus this\npatch implements mentioned fuctionality and can be enabled\nwith a single variable.\n\nProcess is alike to what we have and been using for Octavia for a while.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/784379\nChange-Id: I29984c565c808bca79ef39a02e79d9b015a70786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ce76bece83063bef3f6637e6d550a46d5c92bd7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ce76bece83063bef3f6637e6d550a46d5c92bd7d"}]},"branch":"refs/heads/master"},"3035e2533323e7ab42c75ed65f77ff2f739323c0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-06-04 11:53:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/951590/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/90/951590/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/90/951590/5"}}},"commit":{"parents":[{"commit":"46862920df250163ca624eca834dd9582752ca90","subject":"Do not use group as a \"setup_host\" variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/46862920df250163ca624eca834dd9582752ca90"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-02 19:07:21.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@advanced.host","date":"2025-06-04 11:53:50.000000000","tz":0},"subject":"Add support for generation of SSH keypairs","message":"Add support for generation of SSH keypairs\n\nIf operator needs to enter guest agent instance for troubleshooting,\nthey need to create a keypair, upload it as Trove user specifically,\ncreate a security group and define it\u0027s UUID in trove.conf\n\nThis process is quite cumbersome and time consuming, thus this\npatch implements mentioned fuctionality and can be enabled\nwith a single variable.\n\nProcess is alike to what we have and been using for Octavia for a while.\n\nChange-Id: I29984c565c808bca79ef39a02e79d9b015a70786\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3035e2533323e7ab42c75ed65f77ff2f739323c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/3035e2533323e7ab42c75ed65f77ff2f739323c0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
