)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~991881","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~I8ebe22a611c81d489f0b58f720b25154d8139b84","project":"openstack/openstack-ansible-os_trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-06-15 17:33:27.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-15 17:33:26.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-12 13:42:11.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-06-15 14:10:24.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I8ebe22a611c81d489f0b58f720b25154d8139b84","subject":"Add DHCP egress rule to Trove guest management security group","status":"MERGED","created":"2026-06-05 15:44:47.000000000","updated":"2026-06-15 22:52:49.000000000","submitted":"2026-06-15 22:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991881","meta_rev_id":"6999a5af76c1c93531a3c2f31c71f464faa06f9a","_number":991881,"virtual_id_number":991881,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-15 22:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-06-12 08:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-12 13:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-06-15 14:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-06-15 14:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-06-12 08:26:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-12 13:42:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-05 15:47:15.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-05 15:47:15.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-06-05 19:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-15 13:59:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"5cb5d51c2064f962634ce1d5b235e216ec8007a6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-05 15:44:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8fa80d381e37f89573c26c5a1a1c2c30dcced10","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-05 15:46:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f125755f23c8d3fb5930ca58176f119dfefd4dba","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-05 15:46:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"897bd933fe0658bcb030dac4321c44be5548cd0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 19:25:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a80b5dc08a49978a227cdaeedb3b56\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f3e84f42a5b487da4041d16e33c8b9e : SUCCESS in 8m 18s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/f8b3761a6daf40a4aa897d03a1dc4fb0 : SUCCESS in 1h 36m 01s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/32c389f75c7c4835bfa0f76022e167b9 : SUCCESS in 2h 31m 03s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/18349b9d0fdb43eda5233a855d880aae : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e733ab59fb244eafb868c29b876929b7 : SUCCESS in 1h 38m 34s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e7d4d386c5b480e84ec5b5507d7bbc9 : SUCCESS in 2h 23m 44s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/83c8370f4bbf45e5af679f2cbae0a07d : SUCCESS in 1h 29m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2fdac8f1de495eac0f537f68d38f5f : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0451a785c4a74345ab1816a38d135924 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d450eab24509179e452dd956f669dbdd6cd6ecdf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-07 13:49:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd111d5571d844f09bc6ef641869fc6a568ee4f2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-12 08:26:16.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a21b8ba2dfc697fe4961cd8c92badb9493bc3883","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-12 13:42:11.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a630d2fccede9648502ee6a338be4e9ac84f6b7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-12 13:42:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7fcfd68fe7beb417ba53dd0b0a05b80096fe7ee","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-15 14:00:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3fecf3682088b2b657ffc1939c639fc0c4da66fa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-06-15 14:10:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d05cde98a0c9290051d99b5a8fb73263d12d1c48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 14:14:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89ac6e9f116b445aa4b215291cf7663a83e4a446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 17:32:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/130b2261830948c59e96a5a43ed43951\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ced090fcd04b46689111bab544b4b6a4 : SUCCESS in 14m 32s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/58ab32b311fe455d9768d7e8bbcf7797 : SUCCESS in 1h 53m 18s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/868bdb3967854ec9b39b809e125dcda0 : TIMED_OUT in 3h 11m 37s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b7058b83a0d54b79999a774c63de4410 : SUCCESS in 2h 52m 23s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0938a0f1f078411389dda4acc9cfa1a2 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f32484ad2ece4bb2b51372d1713c3ed1 : SUCCESS in 2h 32m 01s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/887bcd70233a459e8f13a70e4c424b61 : SUCCESS in 2h 23m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f50b61b89e4dc98874295e9a848ed6 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f1343c85e1540cf9a8bb31b95725a87 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a1314b6cdb6e2073d2c7d9032393d4fe283ba4ee","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-15 17:33:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e58692ba65566c48d7c199a93d1e110fed836bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 20:17:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7635977e3d94b92bb63cd673ae81c05\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24f348749b0b4b6ba5479089a868eb8e : SUCCESS in 12m 43s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b3dd9ad635d6490482bcff277b63aa29 : SUCCESS in 1h 29m 24s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/63f7abfe489e4110975f1adb6f040040 : SUCCESS in 2h 29m 39s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/36af9e46ca424676a741cbe3f5a9cc19 : SUCCESS in 2h 32m 39s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2624696a138d4c8b96a7284616410ceb : SUCCESS in 1h 22m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/730073dfeca640c3a90cfa7ef185e865 : SUCCESS in 2h 28m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/dd49a296314e4190a58a816d6e13d3be : SUCCESS in 2h 29m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8127dad4b354f74a35a58fd0b09c3a5 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46a0761044984bacafa90b78c363c701 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":2},{"id":"6213405f59b2ae9691b74ea7df4b7a2383d60fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 20:19:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ba7f85c0be1fee4aa0d980140aad0d4721cbd17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 22:52:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b72a40d0918417e9690ff637b862156\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/377b287e25f84962a86e3bb6953fd50f : SUCCESS in 10m 01s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/83651849f6b944e5b4e8428a5ff3b74e : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3af2c9a41dcd47aaaa34b0127bd5b07b : SUCCESS in 2h 19m 24s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0c79ea7d1ca34592913df045cedea168 : SUCCESS in 2h 21m 48s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/332b8b3966a047ebb5508d4205e965b9 : SUCCESS in 1h 25m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61a2d5f8315f4225a718bc37c545a739 : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ad6abf0230df4f4a8b3f2e207bc53681 : SUCCESS in 1h 44m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecb6bdab32b40e1834a4e0e2c4fe922 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a0d5bb4b9ff4da0980d5e9823205df8 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ef8ad3c9cfc680db75d6b2e79a1246061eafadb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 22:52:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6999a5af76c1c93531a3c2f31c71f464faa06f9a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 22:52:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5ceb3ea0f04810b20117c32b5d78cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8518da3a8bd49b6889d03c08208126a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b210a381c893559128255f5bbeeed66643727635","revisions":{"a91fc354ec7657c7af5175b38ed1889a9ff79cf1":{"kind":"REWORK","_number":1,"created":"2026-06-05 15:44:47.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/81/991881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/81/991881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/1"}}},"commit":{"parents":[{"commit":"908534f8eec082b526075cc46dc346743fe48ffd","subject":"Merge \"Handle missing provider network definition safely\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/908534f8eec082b526075cc46dc346743fe48ffd"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-05 15:44:47.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-05 15:44:47.000000000","tz":0},"subject":"Add DHCP egress rule to Trove guest management security group","message":"Add DHCP egress rule to Trove guest management security group\n\nAllow UDP egress on port 67 in the default Trove guest management\nsecurity group. Without this rule, guest instances may fail to obtain\nan IP address via DHCP.\n\nChange-Id: I8ebe22a611c81d489f0b58f720b25154d8139b84\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a91fc354ec7657c7af5175b38ed1889a9ff79cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a91fc354ec7657c7af5175b38ed1889a9ff79cf1"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"b210a381c893559128255f5bbeeed66643727635":{"kind":"REWORK","_number":2,"created":"2026-06-05 15:46:56.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/81/991881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/81/991881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/81/991881/2"}}},"commit":{"parents":[{"commit":"908534f8eec082b526075cc46dc346743fe48ffd","subject":"Merge \"Handle missing provider network definition safely\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/908534f8eec082b526075cc46dc346743fe48ffd"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-05 15:44:47.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-05 15:46:40.000000000","tz":0},"subject":"Add DHCP egress rule to Trove guest management security group","message":"Add DHCP egress rule to Trove guest management security group\n\nAllow UDP egress on port 67 in the default Trove guest management\nsecurity group. Without this rule, guest instances may fail to obtain\nan IP address via DHCP.\n\nChange-Id: I8ebe22a611c81d489f0b58f720b25154d8139b84\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/b210a381c893559128255f5bbeeed66643727635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/b210a381c893559128255f5bbeeed66643727635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
