)]}'
{"id":"openstack%2Fopenstack-ansible-os_trove~992336","triplet_id":"openstack%2Fopenstack-ansible-os_trove~master~Ia1974d2dfa547f30f109dd02da508d494c142790","project":"openstack/openstack-ansible-os_trove","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-07-03 14:25:53.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-07-03 14:07:57.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-01 14:13:43.000000000","reason":"removed on reply"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-07-03 14:25:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1974d2dfa547f30f109dd02da508d494c142790","subject":"docs: update standalone RabbitMQ guide for Trove","status":"MERGED","created":"2026-06-08 18:32:39.000000000","updated":"2026-07-03 14:27:00.000000000","submitted":"2026-07-03 14:25:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992336","meta_rev_id":"4e16cbc946fe195ced2ea3980505d9db9754ffdf","_number":992336,"virtual_id_number":992336,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-03 14:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-07-01 12:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-01 14:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-07-03 14:07:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-08 21:50:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-10 00:00:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-15 17:33:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-06-15 17:33:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-07-01 14:13:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"6b024ab7c5e94f05df90ac1f003377b42f34e5f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 18:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6f76ec4ff16915aa23c6134889d7ddd90973121","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-08 19:17:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9cd5885ebfeb2824e28e98c58f1611c626d9b2d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 21:50:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3aebe409c1da4c9a9b9dc4fd6af4db38\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cd7ef09b41f4e2a876d53d4e6291a8d : SUCCESS in 2m 24s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8e66894c262408b8ec77c83af52fc1e : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6fe5aba103564c11941e60d2cd204df5 : SUCCESS in 1h 57m 27s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d9579cc56bb44ea18cedea61780c72bb : SUCCESS in 2h 25m 11s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3937c7b893a0446fa93cda9245617325 : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3016fcafceaa4ef09b5c1c5e9738a871 : SUCCESS in 2h 00m 28s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2a2800da66a4c25b8c3e3c404492c52 : SUCCESS in 2h 24m 36s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2ae2bb5935aa4c3e89e7ad8bdca51da3 : SUCCESS in 1h 54m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8e7ae8711d46aeadc554fd8b21ed6e : FAILURE in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f634bd59cb940ea89594ee4e83f8bc2 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"93aeb59a18d5f401cf751dc9ae599b35ad7f2a6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-09 21:53:21.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e5b3cce83f6878bd0ccb76542f47adee27e6bc24","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 23:58:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_37632\u003e","accounts_in_message":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"_revision_number":4},{"id":"554c612532d3448be1782190db8a56582ae489cc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 00:00:44.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f96f673c72ebeb2a61dda70b61d7e949d5d2f792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 00:19:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58cba9d13a7446b381d13ed5f268e118\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38c11f0d47fe4761be34cb5808ec31a6 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6960c4005742aab6d9a6bf6c05ba5b : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b884421cede47f4a8a36ab2e268ad2f : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":4},{"id":"8d5f6bf434592bdabfd1cff121fafea0840b185b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-10 09:30:10.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"14c465b1daa58c1c9374eac4c003e74c2d69674d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-10 09:30:16.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9919aad749df4b622be57886b94bf0949f5b731b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 09:46:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/788075d72f1347998ea106b394a6fb1b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c73efce7663b41fca42dcb6c4601c160 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e156a2d8cf443da0b3d0849344e8cb : FAILURE in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a00cb1ac2e543399db1eb00551533f9 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"e7a13afb0c7bdc317bfb372544ab9e09e6fe800b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-10 10:50:37.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"7af5ee2b4bdab0c0893cd03f1c6266dbb8e6034c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:05:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b93faba9f1745c193818911b5bd6de9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b39e1ecdc14b4348a1cedef80c8e30eb : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07f050cd69a42e0bf25dc91ee7a3d17 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ee1bf598f8f4dc1a251e13f7f3ea0e8 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":6},{"id":"69aaaef7c0259a198b6df33cfa30ce560e951c30","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-10 18:24:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"9d37274a64efbf89e6a56099672294be9b687c58","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-25 07:12:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ad12fed1923e908a98b576a90b9aa3147e7fc27d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-25 23:17:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"9eee7706096bebea58e87180a86f099872d0f4fa","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-26 15:24:08.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"6a5367a9208cef29b5eaee01b73964821ce1745c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-26 15:24:10.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"78ebfed967a275f1f76f1dc5e177cb1052721289","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 15:26:34.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a4521f83344ac635a0bde2bf345ddd0022dfea91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:48:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2912f5de162f4251bda74ef718e65b2b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dc7e52eb11d4286bbcd8549d0e908bf : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f797cce6a729421b93ae59b05b5e5bfb : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2704efa36b8471aa2f30b7a0348cb70 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":7},{"id":"b45b75a846ac57e37034b9280cb6ae95848b83e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-26 15:49:39.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"959dacfd853843b5ddf91926dc7fa27b9f387b0e","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-26 15:49:41.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ba8fe3be2bbf04dc85ec61986e2f38b64ac780e1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 15:54:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f0913995746d1486f32b354f1e1123ce65e9d710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-26 16:02:27.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"aa44bf810661610dff83ce2fe1c853575b77e979","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-26 16:02:34.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d8a1499b9faa9e414365bc28269fc27eaba38234","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-26 16:13:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"97072427a50e89d142ac7107739cd6615205bf7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:14:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4737691a3a2419c9a4bfaca06103dec\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78e12efeb9174ee9ae6e06f7bdf136fb : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8644c23ed82422f822a59cbc4838327 : FAILURE in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b70726f50d134afda1f61d7ae4c8fafb : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":9},{"id":"e7ccdc222d3ac95913db0f985f6458fde9692725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-28 12:08:12.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"7c7fa95d832a45567c83a0c1cc4c2c851027eb4c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-06-28 12:08:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c429bff34cd7276a8dca8a595002b685eaa3685c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 12:23:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95656b0072254c3d82104e639a270447\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2878d1e5cd6414d9a9076d2df78512a : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f56b4d8d2d42e79832295414aa6ea0 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4ba980b1d52491b9d2e5ba7a33948f4 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":10},{"id":"573d346fc21c80cc70b3568f330c81d3de2003ba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-01 12:40:32.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"907eee9bf0c7409c63a0ce7bd111bb8826894f4b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-01 14:13:43.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"186e01af5c4367c1024be65f28fbdc08afa95abb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-07-03 14:07:57.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"243b23343d9f5aa531ce9363f3f1180c204ae572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 14:10:51.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8e38a0a396f3a3e7c73f15e6f074a079539d592d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 14:25:34.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7042e1456e3e415aa70ca633507cb5a6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb188ca8e5fa420b90e3cb2fcd9cb92b : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb36998b3a64a51bfa556ff30eba70f : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01bcdb8a13f49e1a1f42797a2042ff2 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":10},{"id":"8fdafca262f88307794d2912b11f377b2c06e599","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 14:25:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"4e16cbc946fe195ced2ea3980505d9db9754ffdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-03 14:27:00.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a61bee760144c3857611b762c551c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68f1d03fd0347e7881b6e457f21c219 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a434d70da1d58cd27c820ff3431d90efa3adb380","revisions":{"fc3db8d8c3c45d5d475e14ba3a355ba6af5a9916":{"kind":"REWORK","_number":1,"created":"2026-06-08 18:32:39.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/1"}}},"commit":{"parents":[{"commit":"908534f8eec082b526075cc46dc346743fe48ffd","subject":"Merge \"Handle missing provider network definition safely\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/908534f8eec082b526075cc46dc346743fe48ffd"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"subject":"docs: wip","message":"docs: wip\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fc3db8d8c3c45d5d475e14ba3a355ba6af5a9916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fc3db8d8c3c45d5d475e14ba3a355ba6af5a9916"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"fc680916b9856e47e426c17cda9de3a43ac25bd2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-08 19:17:42.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/2"}}},"commit":{"parents":[{"commit":"7865da1caa21fd4a6b5f2b24eb2e0d2de0f041b4","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7865da1caa21fd4a6b5f2b24eb2e0d2de0f041b4"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 19:17:42.000000000","tz":0},"subject":"docs: wip","message":"docs: wip\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fc680916b9856e47e426c17cda9de3a43ac25bd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/fc680916b9856e47e426c17cda9de3a43ac25bd2"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"908534f8eec082b526075cc46dc346743fe48ffd","ours":"fc3db8d8c3c45d5d475e14ba3a355ba6af5a9916","theirs":"7865da1caa21fd4a6b5f2b24eb2e0d2de0f041b4","merge_strategy":"recursive","contains_conflicts":false}},"ecb6202b0df816a3fd6da11085cbe755792aacbb":{"kind":"REWORK","_number":3,"created":"2026-06-09 21:53:21.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/3"}}},"commit":{"parents":[{"commit":"7865da1caa21fd4a6b5f2b24eb2e0d2de0f041b4","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/7865da1caa21fd4a6b5f2b24eb2e0d2de0f041b4"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-09 21:52:51.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ecb6202b0df816a3fd6da11085cbe755792aacbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/ecb6202b0df816a3fd6da11085cbe755792aacbb"}]},"branch":"refs/heads/master"},"03ad4f523e9d24abf3ee3479e8bcf757e1bc8032":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-09 23:58:05.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/992336/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/4"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-09 23:58:05.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/03ad4f523e9d24abf3ee3479e8bcf757e1bc8032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/03ad4f523e9d24abf3ee3479e8bcf757e1bc8032"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"7865da1caa21fd4a6b5f2b24eb2e0d2de0f041b4","ours":"ecb6202b0df816a3fd6da11085cbe755792aacbb","theirs":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","merge_strategy":"recursive","contains_conflicts":false}},"17b04100cfe0276372265046be9b4b3ffd23b429":{"kind":"REWORK","_number":5,"created":"2026-06-10 09:30:10.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/5"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-10 09:29:56.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/17b04100cfe0276372265046be9b4b3ffd23b429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/17b04100cfe0276372265046be9b4b3ffd23b429"}]},"branch":"refs/heads/master"},"170b76ca2f61bd6535dc43d6d1c6246f0eec9623":{"kind":"REWORK","_number":6,"created":"2026-06-10 10:50:37.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/6"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-10 10:50:28.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/170b76ca2f61bd6535dc43d6d1c6246f0eec9623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/170b76ca2f61bd6535dc43d6d1c6246f0eec9623"}]},"branch":"refs/heads/master"},"36e3ec04e9e6cacd19209f6621b4db9a4864560a":{"kind":"REWORK","_number":7,"created":"2026-06-26 15:24:10.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/7"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-26 15:23:46.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/36e3ec04e9e6cacd19209f6621b4db9a4864560a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/36e3ec04e9e6cacd19209f6621b4db9a4864560a"}]},"branch":"refs/heads/master"},"afcf3bf6b566bc8bf13c874975677bea13c3fd1d":{"kind":"REWORK","_number":8,"created":"2026-06-26 15:49:39.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/8"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-26 15:49:30.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/afcf3bf6b566bc8bf13c874975677bea13c3fd1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/afcf3bf6b566bc8bf13c874975677bea13c3fd1d"}]},"branch":"refs/heads/master"},"622e29c585df711d43edaf03504861edef1b7adc":{"kind":"REWORK","_number":9,"created":"2026-06-26 16:02:27.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/9"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-26 16:02:07.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/622e29c585df711d43edaf03504861edef1b7adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/622e29c585df711d43edaf03504861edef1b7adc"}]},"branch":"refs/heads/master"},"a434d70da1d58cd27c820ff3431d90efa3adb380":{"kind":"REWORK","_number":10,"created":"2026-06-28 12:08:12.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/36/992336/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_trove","ref":"refs/changes/36/992336/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_trove refs/changes/36/992336/10"}}},"commit":{"parents":[{"commit":"14f4f4c82f2ddffbf196f87fc304885b8a4ea822","subject":"docs: clarify deployment configuration examples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/14f4f4c82f2ddffbf196f87fc304885b8a4ea822"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-08 18:32:39.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-06-28 12:08:07.000000000","tz":0},"subject":"docs: update standalone RabbitMQ guide for Trove","message":"docs: update standalone RabbitMQ guide for Trove\n\nUpdate the dbaas network binding and add the required RPC\nserver, password overrides so guest\nagents can reach the dedicated cluster.\n\nChange-Id: Ia1974d2dfa547f30f109dd02da508d494c142790\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a434d70da1d58cd27c820ff3431d90efa3adb380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_trove/commit/a434d70da1d58cd27c820ff3431d90efa3adb380"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
